)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~670687","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I359943551cbde52d3f8cf34a56ce21fae4e6b01e","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"I359943551cbde52d3f8cf34a56ce21fae4e6b01e","subject":"[DNM] Connect to the db setup host rather than localhost","status":"ABANDONED","created":"2019-07-13 07:55:13.000000000","updated":"2019-07-27 16:41:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83b7a33365422379c731e5285e9337aeb6a76a01","_number":670687,"virtual_id_number":670687,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-07-18 17:26:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 17:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0cf3297f3ee52e720cc2008b19c5dec27e259e3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-13 07:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2051073851915e8de144f7b37654cd942f9180bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 09:19:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/670687/1/check/openstack-ansible-linters/7b71de1/ : SUCCESS in 7m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_metal-debian-stable/fb8af9b/ : SUCCESS in 55m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_metal-centos-7/3cb77bd/ : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/b688203/ : SUCCESS in 1h 08m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0e6c36d/ : SUCCESS in 58m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d5d0944/ : SUCCESS in 52m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/87/670687/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/9323c64/ : SUCCESS in 51m 53s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/87/670687/1/check/openstack-tox-docs/2222d2b/html/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7148d36d03466c8bdda3749e424e45dc05f58b0a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-14 09:04:17.000000000","message":"Topic set to bind-to-mgmt","accounts_in_message":[],"_revision_number":1},{"id":"404f6a77748bfc2f614407dff189450e3161f77c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-18 14:45:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b4b458391f5e92dbafad7b33e83676a575ad6854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:39:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/87/670687/2/check/openstack-ansible-linters/d43138f/ : FAILURE in 4m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_metal-debian-stable/a9c5a4e/ : FAILURE in 18m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_metal-centos-7/ee75a86/ : FAILURE in 27m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/e876800/ : FAILURE in 22m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/132c8ea/ : FAILURE in 19m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b6ba03e/ : FAILURE in 20m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/87/670687/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/4db5402/ : FAILURE in 20m 56s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/87/670687/2/check/openstack-tox-docs/004dbaa/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d6c768419f14853737bfc1ba9b5dc9aa76333f3b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-18 15:41:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1cc8bd56ff0e143df18dbb8ceb7aaf4c3f1dca03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 17:26:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/670687/3/check/openstack-ansible-linters/c396fcb/ : SUCCESS in 5m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_metal-debian-stable/7f78f0c/ : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_metal-centos-7/95eb69f/ : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/ee15d03/ : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9a9546c/ : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/22641ef/ : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/87/670687/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/f00f2f6/ : FAILURE in 52m 14s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/87/670687/3/check/openstack-tox-docs/2e22949/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"83b7a33365422379c731e5285e9337aeb6a76a01","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-27 16:41:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"95705ff8f86058e5a039a56e2b55b550e758bd3b","revisions":{"ea62c786145b426d81cb8db38c39c9b9ba2c00af":{"kind":"REWORK","_number":1,"created":"2019-07-13 07:55:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/670687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/87/670687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/1"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-13 07:53:26.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-13 07:54:39.000000000","tz":60},"subject":"[DNM] Connect to the db setup host rather than localhost","message":"[DNM] Connect to the db setup host rather than localhost\n\nChange-Id: I359943551cbde52d3f8cf34a56ce21fae4e6b01e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ea62c786145b426d81cb8db38c39c9b9ba2c00af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ea62c786145b426d81cb8db38c39c9b9ba2c00af"}]},"branch":"refs/heads/master"},"5822a8436030de0d884402edc3ee25326acb5efa":{"kind":"REWORK","_number":2,"created":"2019-07-18 14:45:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/670687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/87/670687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/2"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-13 07:53:26.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-18 14:45:30.000000000","tz":0},"subject":"[DNM] Connect to the db setup host rather than localhost","message":"[DNM] Connect to the db setup host rather than localhost\n\nChange-Id: I359943551cbde52d3f8cf34a56ce21fae4e6b01e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/5822a8436030de0d884402edc3ee25326acb5efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/5822a8436030de0d884402edc3ee25326acb5efa"}]},"branch":"refs/heads/master"},"95705ff8f86058e5a039a56e2b55b550e758bd3b":{"kind":"REWORK","_number":3,"created":"2019-07-18 15:41:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/670687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/87/670687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/87/670687/3"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-13 07:53:26.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-18 15:41:34.000000000","tz":0},"subject":"[DNM] Connect to the db setup host rather than localhost","message":"[DNM] Connect to the db setup host rather than localhost\n\nChange-Id: I359943551cbde52d3f8cf34a56ce21fae4e6b01e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/95705ff8f86058e5a039a56e2b55b550e758bd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/95705ff8f86058e5a039a56e2b55b550e758bd3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
