)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~670996","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Ib24a5c96359c349781c48a5d82d7c64e793b8e43","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"Ib24a5c96359c349781c48a5d82d7c64e793b8e43","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-16 10:14:23.000000000","updated":"2019-07-17 06:42:18.000000000","submitted":"2019-07-17 06:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670996-1563345738943-0ae53db0","meta_rev_id":"751b0483e2fed737e4733d9e6ab23f9aca42cff4","_number":670996,"virtual_id_number":670996,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-17 06:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 21:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-17 05:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-07-17 05:15:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 16:13:53.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-16 21:33:45.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-17 05:15:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-17 06:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30613b52838ed0e74230e72df80bbdced03452bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 10:14:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f666dd0f9b8e9d3e1726933deb0543bd1820b78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 11:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/670996/1/check/openstack-ansible-linters/3ad44bf/ : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_metal-debian-stable/6854f16/ : SUCCESS in 1h 01m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_metal-centos-7/fd5f763/ : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/962631a/ : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/447eadb/ : SUCCESS in 49m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4711ba1/ : SUCCESS in 52m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/96/670996/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/744190e/ : SUCCESS in 57m 08s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/96/670996/1/check/openstack-tox-docs/243d44f/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"74b7a7eeb24bfdfaf9cbe9aea024097f9369981b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 12:33:36.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we disable uwsgi logging, it\u0027s duplicating all the stuff the Placement logs anyways?  That way, we don\u0027t have 2 lines for each request","accounts_in_message":[],"_revision_number":1},{"id":"4038466ee488c076933de3c79998e79cd88ba09c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 13:04:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71a0b4d6f21f7e6f8b10dbf80394d0978d53d3f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:03:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/670996/2/check/openstack-ansible-linters/c5d8043/ : SUCCESS in 4m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_metal-debian-stable/7e32f5d/ : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_metal-centos-7/b745472/ : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/5ff1790/ : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8150187/ : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0c9af59/ : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/96/670996/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/df333df/ : SUCCESS in 1h 03m 07s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/96/670996/2/check/openstack-tox-docs/592243a/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4728a1326bbe08b86c8c6d6e0ce0143a24ac3750","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-16 16:13:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"908fe174ca0db89eaabc8789785b1b4699742497","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 19:25:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d62209b7f627edc1dad4698cb998e57125544cc8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:16:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3310008ff94b131a88c3c20946390f39d1f21ed9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:17:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c7b32ff7ff4205e8fcc80a9aa267c0bb891dc5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 21:33:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c78f37beb35fb8c07504bf9fc912a8097e0d549d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 22:51:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/670996/3/check/openstack-ansible-linters/38a2b68/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_metal-debian-stable/9172724/ : SUCCESS in 1h 12m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_metal-centos-7/85461a6/ : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/35b47cb/ : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9cee8d8/ : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6c41e05/ : SUCCESS in 59m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/96/670996/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7a27df1/ : SUCCESS in 1h 00m 28s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/96/670996/3/check/openstack-tox-docs/ab14107/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2def77d615893befdc152b29fabf6de03c6fc76b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-17 05:15:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"26d8f7e0188271684efbf37a58ed61332d61e454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:16:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"178b2a06c7bf6b25ee30ce04506c113d0fd5f4a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:42:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f128300a32d75ed7d1cd8d65376e1f22ad64e85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:42:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/670996/3/gate/openstack-ansible-linters/d47fe1a/ : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/670996/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/187dbb0/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/670996/3/gate/openstack-ansible-deploy-aio_metal-centos-7/5b78382/ : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/96/670996/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/59fce95/ : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/670996/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/81c38dc/ : SUCCESS in 56m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/670996/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/fcd760e/ : SUCCESS in 57m 39s\n- openstack-tox-docs http://logs.openstack.org/96/670996/3/gate/openstack-tox-docs/6d89245/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab692dbc56f0651cfafa3b2993e6a47cc7c8e736","revisions":{"7577b790b3918aba88db5367af6793cebd3a2371":{"kind":"REWORK","_number":1,"created":"2019-07-16 10:14:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/670996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/96/670996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/1"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 10:13:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 10:14:21.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: Ib24a5c96359c349781c48a5d82d7c64e793b8e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/7577b790b3918aba88db5367af6793cebd3a2371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/7577b790b3918aba88db5367af6793cebd3a2371"}]},"branch":"refs/heads/master"},"4df2a945b8dd374e92720fb05cedca2055de8a95":{"kind":"REWORK","_number":2,"created":"2019-07-16 13:04:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/670996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/96/670996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/2"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 10:13:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 13:04:10.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: Ib24a5c96359c349781c48a5d82d7c64e793b8e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/4df2a945b8dd374e92720fb05cedca2055de8a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/4df2a945b8dd374e92720fb05cedca2055de8a95"}]},"branch":"refs/heads/master"},"ab692dbc56f0651cfafa3b2993e6a47cc7c8e736":{"kind":"REWORK","_number":3,"created":"2019-07-16 21:16:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/670996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/96/670996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/96/670996/3"}}},"commit":{"parents":[{"commit":"77e709668675c02e8f988b36bb4b1eea94913436","subject":"Fix deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/77e709668675c02e8f988b36bb4b1eea94913436"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 10:13:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 21:16:26.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: Ib24a5c96359c349781c48a5d82d7c64e793b8e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ab692dbc56f0651cfafa3b2993e6a47cc7c8e736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ab692dbc56f0651cfafa3b2993e6a47cc7c8e736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
