)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~673761","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I342dd945b823b2b8134d759df2e2e2dd06a74ad4","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"example_playbook","hashtags":[],"change_id":"I342dd945b823b2b8134d759df2e2e2dd06a74ad4","subject":"Add example playbbok","status":"MERGED","created":"2019-07-31 08:43:30.000000000","updated":"2019-07-31 16:32:31.000000000","submitted":"2019-07-31 16:32:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673761-1564590751178-05bf691b","meta_rev_id":"145490b34b745e2915518039c7c3cf6055971c9d","_number":673761,"virtual_id_number":673761,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-31 16:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-31 12:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-31 14:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-31 14:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 12:59:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-31 14:43:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-31 16:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"335e3d647b25a8d3cee6c52f17e10e8bd3a34bc3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-31 08:43:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"283430b8baa90696304f7c5593217c0bf21645a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/673761/1/check/openstack-ansible-linters/2d6b9b4/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5439b92/ : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_metal-centos-7/c343ae8/ : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/ba9f3f1/ : SUCCESS in 1h 14m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/aa94379/ : SUCCESS in 1h 01m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0ba01f3/ : SUCCESS in 1h 01m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/befdd84/ : SUCCESS in 1h 03m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/61/673761/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/9bced6c/ : SUCCESS in 59m 56s\n- openstack-tox-docs http://logs.openstack.org/61/673761/1/check/openstack-tox-docs/13a1e4d/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1a5d2dc7c3955b0b88d09a29915893acd7e92d1a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-31 12:59:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81b74c941645e90473156cb50f0bbe88657870ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-31 14:43:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce4468a8fd9d652136cc48b4cb823b068fb50b6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 14:43:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c02d1c7b5f3e674d3251b4ffcc2c06a4b9d79551","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 16:32:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/673761/1/gate/openstack-ansible-linters/056aee2/ : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/f397b3c/ : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_metal-centos-7/fdbc7f6/ : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/9ca41f5/ : SUCCESS in 1h 28m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9ca1c45/ : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/1d6b82d/ : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/61/673761/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/04d1663/ : SUCCESS in 44m 37s\n- openstack-tox-docs http://logs.openstack.org/61/673761/1/gate/openstack-tox-docs/18a688e/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e57fba91cb234b2751c57402e50c0c8a1abb7359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 16:32:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67f77de713324a8b83dffb9d5d3da5e05d6dc6a8","revisions":{"67f77de713324a8b83dffb9d5d3da5e05d6dc6a8":{"kind":"REWORK","_number":1,"created":"2019-07-31 08:43:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/673761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/61/673761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/673761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/673761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/673761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/673761/1"}}},"commit":{"parents":[{"commit":"d9e04c6c08959f43c7d6f90599f3c73b1280ee7b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/d9e04c6c08959f43c7d6f90599f3c73b1280ee7b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-31 08:43:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-31 08:43:15.000000000","tz":180},"subject":"Add example playbbok","message":"Add example playbbok\n\nAdd playbook as an example, since it was included in docs,\nbut was empty.\n\nChange-Id: I342dd945b823b2b8134d759df2e2e2dd06a74ad4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/67f77de713324a8b83dffb9d5d3da5e05d6dc6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/67f77de713324a8b83dffb9d5d3da5e05d6dc6a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
