)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~675161","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I9c222cfc28c6838099bc155bf2d023cfd6112816","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"static-imports","hashtags":[],"change_id":"I9c222cfc28c6838099bc155bf2d023cfd6112816","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-08-07 17:27:39.000000000","updated":"2019-08-08 14:22:22.000000000","submitted":"2019-08-08 14:22:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675161-1565274142486-01c9335f","meta_rev_id":"f1c1ba9ab2824093991528424c328e729620919c","_number":675161,"virtual_id_number":675161,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-08 14:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-08-08 12:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-07 21:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-08-08 12:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 21:18:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-08 12:47:13.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-08 14:22:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95573ae890e690a2c0580560d24afff2619e0b5a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 17:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7976cba307860dab7919f85f1ad6e568c1ecd645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/61/675161/1/check/openstack-ansible-linters/4cb0eef/ : SUCCESS in 8m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_metal-debian-stable/d472e23/ : SUCCESS in 1h 07m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_metal-centos-7/b9d1620/ : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/569fb55/ : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6587367/ : SUCCESS in 1h 07m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/28fe5d7/ : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/aecca67/ : SUCCESS in 58m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/61/675161/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/6390d1a/ : SUCCESS in 53m 11s\n- openstack-tox-docs https://logs.opendev.org/61/675161/1/check/openstack-tox-docs/eab32f5/html/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d8084c37296b5c26d15d684c2baa4f69d91d3407","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-07 21:18:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60c8cab5115a1f857de4fd93b15ec8f963092971","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-08 12:47:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2010cb7609e96cc89427dcfbec26200c758e48b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:47:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad9b035101738f4f216647fffc723a382d5b21b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:22:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cff996db6fc6958f0dddcc59adae0c88d185ad6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:22:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/61/675161/1/gate/openstack-ansible-linters/4fc163e/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/1db8949/ : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_metal-centos-7/41f5b32/ : SUCCESS in 1h 28m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/932e057/ : SUCCESS in 1h 06m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2fdc03e/ : SUCCESS in 1h 06m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/9cd126e/ : SUCCESS in 58m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/61/675161/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/5229c7d/ : SUCCESS in 51m 39s\n- openstack-tox-docs https://logs.opendev.org/61/675161/1/gate/openstack-tox-docs/ae61515/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b561dbf67af61ff07543b469000d3692f4f9184d","revisions":{"b561dbf67af61ff07543b469000d3692f4f9184d":{"kind":"REWORK","_number":1,"created":"2019-08-07 17:27:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/675161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/61/675161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/675161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/675161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/675161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/675161/1"}}},"commit":{"parents":[{"commit":"67f77de713324a8b83dffb9d5d3da5e05d6dc6a8","subject":"Add example playbbok","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/67f77de713324a8b83dffb9d5d3da5e05d6dc6a8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 17:27:29.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 17:27:29.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I9c222cfc28c6838099bc155bf2d023cfd6112816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b561dbf67af61ff07543b469000d3692f4f9184d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b561dbf67af61ff07543b469000d3692f4f9184d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
