)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~683123","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I7b30e575720df62a6e2eaf5f865be56c9de09972","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"osa_leap15_python3","hashtags":[],"change_id":"I7b30e575720df62a6e2eaf5f865be56c9de09972","subject":"Fix distro install on openSUSE","status":"MERGED","created":"2019-09-19 12:16:39.000000000","updated":"2019-09-20 22:03:50.000000000","submitted":"2019-09-20 22:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683123-1569016967539-0f18a21d","meta_rev_id":"52681e998fe62a8788c711576b278bd2a5defb86","_number":683123,"virtual_id_number":683123,"owner":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-20 22:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-20 17:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 19:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-20 19:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 17:37:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-20 19:52:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-20 22:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63b6f881a9ebe4363ae8b22a9a8795dd165fe33a","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-19 12:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"682b9e81e419fdf481a9f1163889c3c2a80951eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d9a9bf15e1f646d89087e0ccd49a9f5d : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ee9c62c0841c4b84b32b7ad59ab8919e : SUCCESS in 1h 25m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5ab454774ea4a62a8c6750504cfa0b5 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4e0957bc0c1f4c37811955c3fc34a7ae : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fea6682724f54bfc96c29531187b3888 : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e608e46ce6c43458dfe3874246a5142 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba5c8f1925364eacb5478cdfbf2ee80b : FAILURE in 34m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d618d228f5594492a40dc815dad604cd : SUCCESS in 1h 05m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d6a32048e3431e9e6bf7ce30ed0557 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"92405ad613382cb1a5d4908fc1081abfc761fb48","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-19 15:43:54.000000000","message":"Topic set to osa_leap15_python3","accounts_in_message":[],"_revision_number":1},{"id":"61c9a0b3850790dc4da2c90467c5e5ee72d575ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 17:37:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46c915694f3123a13c1806f98b387522b3543c00","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 17:57:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"723d8023b967b391d9836fe45b238bf55381cd3e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 19:52:49.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"55a5d92bf656820cba948fb565936c8d2f4dd25c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 19:52:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a2fdb12ce89f4af80919a3a094248bc56de555c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:53:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb9fe81cdb3271230570f92f570ca1623efbe4d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3caa9fb8792946e7bb866fbd96b43671 : SUCCESS in 9m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/da8e9775920c426dae6f69a399bc3f96 : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b0d872e15db4b1dabcc1eeb5cc198f3 : SUCCESS in 1h 42m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1a4bfd23cd6f43319c96ece1e3c051c4 : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4b63f938b51452e995f49defec5c4ae : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5809bae32f9e43778e4f132cc1c8e225 : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20983e6603204952abc9e4f72da60b7b : SUCCESS in 51m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e08b436af44a658762cce5b8a5a0ad : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a34703d7511709a7a1abd87d24b6a1181e9719e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:02:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52681e998fe62a8788c711576b278bd2a5defb86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e091bd4c7d3a408c878553e70a3716e6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2223ab1eadfeb8a9b3e5da9f9f71ababf5dc576","revisions":{"a2223ab1eadfeb8a9b3e5da9f9f71ababf5dc576":{"kind":"REWORK","_number":1,"created":"2019-09-19 12:16:39.000000000","uploader":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"ref":"refs/changes/23/683123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/23/683123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/23/683123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/23/683123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/23/683123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/23/683123/1"}}},"commit":{"parents":[{"commit":"86bca7585387a5c432eb0ad489d231c31a909d9f","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/86bca7585387a5c432eb0ad489d231c31a909d9f"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.com","date":"2019-09-19 12:13:32.000000000","tz":120},"committer":{"name":"Ralf Haferkamp","email":"rhafer@suse.com","date":"2019-09-19 12:13:32.000000000","tz":120},"subject":"Fix distro install on openSUSE","message":"Fix distro install on openSUSE\n\nWith Train openSUSE is switching the openstack packages to use Python 3.\nThis means that we need to use the python3 variants of python-module\npackages.\n\nDepends-On: https://review.opendev.org/683120\nChange-Id: I7b30e575720df62a6e2eaf5f865be56c9de09972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/a2223ab1eadfeb8a9b3e5da9f9f71ababf5dc576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/a2223ab1eadfeb8a9b3e5da9f9f71ababf5dc576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
