)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~711717","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I9191da7a143ab05266d03f1d6ac485c6236d1d42","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I9191da7a143ab05266d03f1d6ac485c6236d1d42","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-03-06 17:58:48.000000000","updated":"2020-03-07 22:57:11.000000000","submitted":"2020-03-07 22:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711717-1583621761117-3b0434b8","meta_rev_id":"79e94be27055b869dfe77d74501fef5925daa8de","_number":711717,"virtual_id_number":711717,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 22:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-06 20:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-06 20:02:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 20:02:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-07 22:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee8138690d796e99301d3523f880fa989465b360","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-06 17:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5b5754c880ad9ecd73e0f2d9fd6a3586009c7bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:01:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a8bb05d158434e8bb7c37fa15ac5cb16 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/68c34930b9564208b8bbf1fc456bbdad : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4f3f259b560d45d2b32e5b9556cc3891 : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c45de2e9cb8f48e893305db79ab2af25 : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/101fcdd4c95848feb9e75e3b697d04b6 : SUCCESS in 1h 07m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8c82e71e5bf743c8bcee422ceadac529 : FAILURE in 1h 30m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08ddb152f770410e8880b6601d49f5d9 : FAILURE in 1h 01m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2e9d3311159f43e7914f69990656c2c0 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e05cd3f6e85e4f52acb90996053990ea : FAILURE in 31m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19cc043ecafa463ebfd2e4a109aa8a43 : SUCCESS in 59m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36379f22fa9544af87c21e90dc3bcaf4 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e47ce1fde53f43c89da5bb376d086df3 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3292ebfea2f1e0edcb48a70d47354e744db233e4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-06 20:02:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d16bc924ec277eecafb4a4ad801f50eeacf407c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:02:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23bca485aff6f7ad9d9b8224384f4840e1895526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 21:39:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7e74587d49ea4f9e8e7f7bea9c8d5f86 : SUCCESS in 5m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/be129eb9f67e4b96831e28729cf1fa15 : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f39255eee9c84745b4d14d3b25e2cdf7 : SUCCESS in 1h 25m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4f7139368a0d4cf2987b551dbcddd868 : FAILURE in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c944003db95487ab333de4af50844c1 : SUCCESS in 1h 35m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5192e04cd544bf4a6ffacb7a5e37826 : SUCCESS in 1h 10m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc5c37eea95242bbbc8efd6d6dc82da4 : SUCCESS in 56m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ac66a4dc9043d8b2d1d4aca269dbfa : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73184c0aad6e4b7583632276d31b9adf : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"eefa88c317519a0c0f96d8c2584c048114491417","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-07 19:49:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c734609e33fdfb42db407320cc0dc7d451172e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 21:30:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6a44463b015f400682d5c149dc6aaa24 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8d6a32555c724149abf1eb67a6b15854 : SUCCESS in 1h 02m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f33ee4b06a7e4a898ed8e8ba5af2aa4f : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6509c970e7e54b5fa20ffbdd1bf62f2d : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23f5a63815734181bd080dc4d55d8615 : SUCCESS in 1h 18m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c46cf72818314aeeab85108049d4b67c : FAILURE in 1h 21m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0888fa0fc91544d79b28f5ab2b1b9568 : FAILURE in 1h 00m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6f033bced444eeab29129b5eddc21d3 : SUCCESS in 1h 04m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/42db81445c8042038a99a378d7a99037 : FAILURE in 40m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92669394c5034777bd191d7336e60183 : SUCCESS in 59m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de7adae268b4d2c8398292c39ef19e8 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9b331be3ba3420994f796178d268d23 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9b4554961ed4cd418e899bae13afc96ab841f93b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 21:30:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e34b06b2b4bcffe393417a27f48a9b329d94e55d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 22:56:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1f79a5c31cba47aa99ab7cc34509d64e : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/03f1256fe5564b08a9e9314ec89acb69 : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1c92a83159de4936b47761b9e0750020 : SUCCESS in 1h 25m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1eb2128e0e4e497cb68c3c697d613731 : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a17621fbb4c34d9c91ab89defef3138b : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6805656430e74f9c8beab81d0e1daa33 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7e7fbba1435418d89cc58aff60ade2e : SUCCESS in 55m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a112a6b922945409009204349bbc24c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66d9d9cd5192407db2a8dc57d0bdf834 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a56ebe539eec585c77babfd66dae229c74ed5449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 22:56:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af602d01d67dbfaad647adfb07a4511ac13bbe5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 22:57:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864b4523db774ef58d562539df9c86fb : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0e15d29c9e6403fb3a6f7dd5a22773b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"284c725ed824d5dce24f1ab5f4d06a3e485fdbc2","revisions":{"284c725ed824d5dce24f1ab5f4d06a3e485fdbc2":{"kind":"REWORK","_number":1,"created":"2020-03-06 17:58:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/711717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/17/711717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/17/711717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/17/711717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/17/711717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/17/711717/1"}}},"commit":{"parents":[{"commit":"ece8744dd68acb6edea84c2f97520f7a16acff80","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ece8744dd68acb6edea84c2f97520f7a16acff80"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:58:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:58:46.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I9191da7a143ab05266d03f1d6ac485c6236d1d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/284c725ed824d5dce24f1ab5f4d06a3e485fdbc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/284c725ed824d5dce24f1ab5f4d06a3e485fdbc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
