)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~745861","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~I0e6a8b1d8b5da0299060e3f21698aa81ec7d7ce4","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I0e6a8b1d8b5da0299060e3f21698aa81ec7d7ce4","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-08-12 11:35:02.000000000","updated":"2020-08-14 10:35:25.000000000","submitted":"2020-08-14 10:34:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745861-1597401240996-2e16d46c","meta_rev_id":"2e9bb86af32998dbd7d27f785b1888fb4700f3b7","_number":745861,"virtual_id_number":745861,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-14 06:11:43.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 10:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 05:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 05:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 05:41:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-14 06:11:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-14 10:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fe3acce3f377274b46c968dd9a446bb4075fc98","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-12 11:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37f3ad0b29e98276dce2114ec9f47a101191c045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:06:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/053e1aeb498847089621f1aa3b102923 : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d5db418375f44bafb403cc6d6df82e02 : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1c010fad21d46ae85bbe05e7d2f7de6 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7527b8c7d47e4f9899f03dbf3451a3eb : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbf29057c3a94449931473e166e0dd54 : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a76c453851a4b5a937b188d34ba890b : SUCCESS in 1h 06m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a9d985fcecb34884a3b8c2e99e2a297a : SUCCESS in 2h 16m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27df9b0e20874b788b29f49ab36eb9f5 : SUCCESS in 1h 48m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6dbe8f0431294f4097cb4632e57a0eb7 : SUCCESS in 57m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/89a67500221848558cbac8fa5e7d4c94 : SUCCESS in 1h 02m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4f02c329d0341faa63f7ca7f502bd6b : SUCCESS in 58m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23a0f9128184f51b2e361e677469e67 : SUCCESS in 50m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1b1fed5a9a43f88b6677a8b194f0d0 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b596ca697c89d6b6e6db8e3e111e646408fb1b7e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-13 05:41:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"424f091f5431ccd131130129436c178c885d280e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 05:41:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0342c0ae0378d5ced0f8f9947e2b834e5389966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 09:00:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb22ee46c5034ab6b6a3992efaece5bc : FAILURE in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4f266bfef98d4fe8ae61f062e3ddc361 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac27b5e9ffc845c08568cc7ae6360d26 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b60f566589b4124945b8c0b8d400fab : FAILURE in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d225aee321b49ae90fe8fb08301fe93 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/534d88afaadb4fd7a3195edcd7ab9a06 : SUCCESS in 1h 04m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fc8bc399762d459c98db01aff7ba66b4 : SUCCESS in 45m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/640cc6efb72c4942b3233fb963657773 : FAILURE in 1h 14m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1d19818cafa4a77bc4b85a3350fbebd : FAILURE in 1h 59m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7effda52fd5415f8442e40e4c38587a : RETRY_LIMIT in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5552447dcc1e4228b43c1041b8719298 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7d503289390f560e8f6a95f46460537fbc50d5c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 06:11:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"077542ed54013ba7fc7bf946df32f7a3dbd79610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 08:50:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ae8473a79741413e94aafb84a2b56239 : SUCCESS in 8m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/460ac57686b04c4da1ed30867b432f6c : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83f05851175b4a75b011b08e755c60e4 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2c43ceddeeb64ffa8aaab7b36bb02a1b : SUCCESS in 1h 21m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6332b5b8b4504360b989683555cc0ab1 : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/279aef180de74c1ab8cca0b7d9272a73 : SUCCESS in 1h 00m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a97444e938d4cd1b8f563d8047502ec : SUCCESS in 2h 19m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b23672d99d04542820c4858e6692fe6 : SUCCESS in 2h 00m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e41b5a804a43461ab8e6247d50d30db8 : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9bea173920d543268662b285f01faf27 : SUCCESS in 59m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d21c2ea471e84c7eae32fc56ccd415b0 : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e176c5ddb734ca0840430db87470647 : SUCCESS in 52m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85194bef6c464667aaccfdf8f8ceb17f : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"06ae8bdc85f5a56e67b117aaf5f15c1a0cbdce82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 08:50:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b373f3be70f636ddd024517426b9e20e6c28631","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:34:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c0f28f421786383eedcd653bc0bf7c69aa1b52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:34:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/18fb80f00b4b48f7b170fa62dc9f0472 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9ca85d4923ae4cd09192f2903b875dad : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5d10e96659c4ddeb1c595970c0456fa : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e418c7b16c34e7195ca0e5f591a5ea2 : SUCCESS in 1h 36m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28524e422c2546268e125f8c52a34678 : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e170d24d364f82a1382d38a51b5a01 : SUCCESS in 1h 02m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a31b90fe4c434fb18e1d86a20f56a9d3 : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dfd293c6378b461cb3078c7912ea078d : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19bb6073f234485d953607d7a13285a0 : SUCCESS in 58m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e6a71ef2bd44248d463022b44dd18b : SUCCESS in 52m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8e495681dc421f8ca006d9674136c5 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c062e371db9aa527dd228ca64a1bbbfb4ffd7637","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:35:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a62322fe9a6485bbd7b02ac0c8c94c1 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce3419f328b71219530f7edf19ac2eaca8b653bf","revisions":{"ce3419f328b71219530f7edf19ac2eaca8b653bf":{"kind":"REWORK","_number":1,"created":"2020-08-12 11:35:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/745861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/61/745861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/745861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/745861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/745861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/61/745861/1"}}},"commit":{"parents":[{"commit":"8a134fd493fd858cb4ded1cf9beae049e8b6c53e","subject":"Merge \"Add Centos-8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/8a134fd493fd858cb4ded1cf9beae049e8b6c53e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:35:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:35:00.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I0e6a8b1d8b5da0299060e3f21698aa81ec7d7ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ce3419f328b71219530f7edf19ac2eaca8b653bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/ce3419f328b71219530f7edf19ac2eaca8b653bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
