)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~880036","triplet_id":"openstack%2Fopenstack-ansible-os_placement~master~Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125","project":"openstack/openstack-ansible-os_placement","branch":"master","topic":"osa/systemd_restart_on_unit_change","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-11 20:26:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-04-11 10:42:02.000000000","updated":"2023-04-11 20:27:49.000000000","submitted":"2023-04-11 20:26:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880036-osa/systemd_restart_on_unit_change","meta_rev_id":"cff5effb02d7246ea9e37ffbb3cf28782ec52614","_number":880036,"virtual_id_number":880036,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 20:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-04-11 18:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 15:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-04-11 18:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-11 13:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 15:11:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-11 18:09:42.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"fa01518be1bf3aa3433954ece0ffcd6a0f815f0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-11 10:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60ed59093f015db97578ea50d4de84655b759075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 13:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b53a6e0014240a59197495e33aeb5a8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca3d1767648f4ffe8e63c96386fbcf75 : SUCCESS in 12m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1cbba5c44bd44709ac0767050869226c : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbc21d4cc0a84dfbadd75297f95f756c : SUCCESS in 1h 34m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c6c84eb78ab4efebad7e94359342502 : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d4fc1dfedd4212b851379cfbfdd8b0 : SUCCESS in 1h 43m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3ed90fc8c2d4a06b0e0a1dfb54aaa59 : SUCCESS in 1h 31m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83eeb69f277d465dbe5351a8638525b3 : SUCCESS in 2h 16m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf706386325f4b73a4b91dff5ecf7fec : SUCCESS in 2h 14m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52b83fe75aff424dba8253a00f0293cd : FAILURE in 30m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6592ac3c56f74fb88b3376606b388c26 : FAILURE in 28m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/999e6f31a39640ae99168e24f2ab1018 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0481e69942cef50484c7075ccf7fe093900462a5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-11 15:11:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbb2fa994f8312371cf42cc6837a80b06e821aae","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-04-11 18:09:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27cf7724da79a82c903a89049476d4405567ad3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 18:12:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ce08cf01129e79572d895e0687e1f4fd6966f59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:26:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df97ef30d85c4fdf9b27f1e98369fa52\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38086123c784842b5812ebe0bfeb53c : SUCCESS in 11m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9722eaa01c15413294b41ae939c85cfd : SUCCESS in 2h 06m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41feb682344347d5ba3d2010ee22c647 : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c6182bff38f43c380c050427039f440 : SUCCESS in 1h 44m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce7ef8a53ba42a48859ae733cb75bdb : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d758cccbd57500f8d43be36b557f08d7b7df0e24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:26:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cff5effb02d7246ea9e37ffbb3cf28782ec52614","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:27:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bfadda2006742ddb19a83896ff19ec8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02e58d954d6408488fa8dba9a8f1a91 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ad5bc1d84f245fe59f5c52ac5227e4cabff6424","revisions":{"5ad5bc1d84f245fe59f5c52ac5227e4cabff6424":{"kind":"REWORK","_number":1,"created":"2023-04-11 10:42:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/880036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/36/880036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/36/880036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/36/880036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/36/880036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/36/880036/1"}}},"commit":{"parents":[{"commit":"250411c408af13d07e67a4d8148937fdc812a6ce","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/250411c408af13d07e67a4d8148937fdc812a6ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-11 10:41:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-11 10:41:51.000000000","tz":120},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/5ad5bc1d84f245fe59f5c52ac5227e4cabff6424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/5ad5bc1d84f245fe59f5c52ac5227e4cabff6424"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
