)]}'
{"id":"openstack%2Fopenstack-ansible-os_placement~882009","triplet_id":"openstack%2Fopenstack-ansible-os_placement~stable%2Fyoga~Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125","project":"openstack/openstack-ansible-os_placement","branch":"stable/yoga","topic":"cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-05-03 10:53:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-05-02 11:02:01.000000000","updated":"2023-05-03 15:01:02.000000000","submitted":"2023-05-03 15:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882009-cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","cherry_pick_of_change":880036,"cherry_pick_of_patch_set":1,"meta_rev_id":"1892bd6a5daf0baa2862cca5446f48a755b2fc2b","_number":882009,"virtual_id_number":882009,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 15:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-02 17:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-02 17:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 14:05:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0b05e230c4ba47454c1fc3197522dd081bc8069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:02:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8c01369158cb76df61c6a00ad758a972d0f486d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:05:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb357d116d9407da0b65373c034c38b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76251aae6ee04bc6a7f9487550546855 : SUCCESS in 11m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/caeec064405645738fd3cf4a47825cb6 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3687707639aa48cc9ee5627d9b431451 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c12989f9254485aa157e32c7419a3a4 : SUCCESS in 1h 25m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f93b5c72a234a63864d7b697622ae05 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a23509e8a7744469126b526df772c1b : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e648a25b01c84e978c4fea877cc399fc : SUCCESS in 1h 14m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f9821e0f94f5438c93e868337f188c8f : SUCCESS in 1h 40m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/153ba8b6791f416e94b91c11e1eeaacd : SUCCESS in 2h 09m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe655594fe8f41fda1b269769b343b54 : SUCCESS in 1h 57m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d91814b99a22460fbb40da0fb9469e09 : FAILURE in 1h 05m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1565e33f2454dc393ff412236491e37 : FAILURE in 41m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de6f222df6524a1cac73e08f08e52205 : SUCCESS in 1h 15m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c67bb55f81c48b7a882724d234e4cb9 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cf378ecf0dc1829ff251ff13995a051dded45438","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 17:41:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0436310389765aba5b25b869f0184a6f8ad686cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 17:43:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b65fc09fb5360555020eff2d53d33293821622b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:01:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/265018f9013e45a48b1aea1ba837bf7e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/014cddf97d4a47209872d07e95d7e07b : SUCCESS in 12m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/febe996f953b425ab02532b70bea4218 : FAILURE in 1h 12m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/01b5b6260d834189b167734be00d76cc : SUCCESS in 1h 58m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1928dcd40be442839399b95526cd9ed9 : FAILURE in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37daf25555934ee98de4a14e9448b6d2 : SUCCESS in 1h 28m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/bd5eeef83e7d4810b48cf424aa74cc4b : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc091311b99e43198647fa33fa480a24 : SUCCESS in 50m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aec63d3e3ddc4039a2ccf3b5e90815f9 : FAILURE in 41m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5d4b17ddb740338b708ccdc561234c : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e3417a4212733b2614cb58a142a21cca04fe3467","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-03 10:53:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9b8ffb3c6d56d8a12508efd3cfed7727ce8d295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 13:10:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9485b9db48bf42689f7a2013f8e5089b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/288e8f8c8f394e10b029717c779c87b9 : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ba50fe2a9ae467585efd0b12028c140 : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/31aa88247415422793166628444cf4c8 : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cea50e95211745ce9f192b961bb94e7d : SUCCESS in 1h 24m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8895d282720b4c318c2523c91cd91154 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39528ebbc33b4994aab7ed5de6d9cd76 : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d27151cfd7254612828fbc9bad015e10 : SUCCESS in 1h 27m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7a5a148467fb4d93ae3b121bee8bdf2c : SUCCESS in 1h 24m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/084b1f6ddda6457692a9feb08ed84588 : SUCCESS in 2h 13m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53018133d9fb4c6b91067f614b7bb812 : SUCCESS in 2h 06m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ec13a96356749c1963abab4b4f99aad : FAILURE in 1h 03m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d6a0fc5a6e9b4629a0398a322634ba62 : FAILURE in 41m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15642b6d9f4046c698b238c73eaf71f6 : SUCCESS in 58m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc4d78e697d495db8ede71a590df9ca : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"006130fc84051f67fbefec6178ecec49731a1898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 13:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de1d0b3a9b100ba123f04fd5da9dff0780b6e0d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:00:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44daaaa714394c30a01a64ecdd446108\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3fc75f24d3f47d7970eab9335ae874e : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/65551cfbeafa4b2f9c1b38bbd5ef2f34 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/16c9c2fb39714aab9465069828703030 : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f97b592a22c6486b89ab6ddf102f8767 : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb784a2dc53042869dd899887d34a926 : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/36c5fd7c675a4c249e3dfa3491058887 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/094b2bb73510463ea3c301420226d45d : SUCCESS in 57m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad086cc136884a959c0531ad0344c45b : FAILURE in 39m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f5407dfbb44760806b216acd5713d4 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b661e47eadb22274644a79b8caaca0dc25a1113f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:00:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1892bd6a5daf0baa2862cca5446f48a755b2fc2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c8e35c1455e42e8bff959fd79130379\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645bc362a6404bb294921dfc1174cbad : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b28d0baa97f2d94ba43c9c8f4a333a7069123fb6","revisions":{"b28d0baa97f2d94ba43c9c8f4a333a7069123fb6":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:02:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/09/882009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_placement","ref":"refs/changes/09/882009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/09/882009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/09/882009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/09/882009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_placement refs/changes/09/882009/1"}}},"commit":{"parents":[{"commit":"bc9773f0b62c95b0654d3e41d424a8e380107b96","subject":"Install git into placement containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/bc9773f0b62c95b0654d3e41d424a8e380107b96"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-11 10:41:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:02:01.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: Iaa2f286fa25412a1ed8c72c0b62bacd3ac50f125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b28d0baa97f2d94ba43c9c8f4a333a7069123fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_placement/commit/b28d0baa97f2d94ba43c9c8f4a333a7069123fb6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
