)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~654375","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Ie11d5dcf5912f23684ebd93ff0548791d57bded6","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie11d5dcf5912f23684ebd93ff0548791d57bded6","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:15:57.000000000","updated":"2019-04-22 12:26:56.000000000","submitted":"2019-04-22 12:26:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654375-1555936016142-eaa21e4c","meta_rev_id":"7ce1d906241eece0ca4ca6af9ee58044dfea463b","_number":654375,"virtual_id_number":654375,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:54:09.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:54:09.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:54:09.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fab06786f8d489674415ca7105f95efba439e923","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4877dc6738fb71cd83d4efe556c70f260316d6a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:45:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/75/654375/1/check/openstack-ansible-linters/acd23c5/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/75/654375/1/check/openstack-tox-docs/a137a89/html/ : SUCCESS in 2m 39s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/75/654375/1/check/openstack-ansible-deploy-aio_metal_rally-centos-7/ac0b47a/ : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/75/654375/1/check/openstack-ansible-deploy-aio_metal_rally-debian-stable/dd3f1a2/ : SUCCESS in 52m 54s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/75/654375/1/check/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/507a5b4/ : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2fba68aaff88f45fcb917fe920b2bfa37ab73502","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:54:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3cdc2b8bcb1d9a321860220378da0973100bd74a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:54:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecbe89b7611cd9808164be4cd89546ab90678dfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:26:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/75/654375/1/gate/openstack-ansible-linters/05e425a/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/75/654375/1/gate/openstack-tox-docs/6a34a4e/html/ : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/75/654375/1/gate/openstack-ansible-deploy-aio_metal_rally-centos-7/d117672/ : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/75/654375/1/gate/openstack-ansible-deploy-aio_metal_rally-debian-stable/5e0a618/ : SUCCESS in 54m 28s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/75/654375/1/gate/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/4afd4a2/ : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7edc2074cf62d9065c1668f2c441e78c5be55418","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:26:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7cdb5c3ea34441015275915541a7ec671f9cd2c","revisions":{"b7cdb5c3ea34441015275915541a7ec671f9cd2c":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:15:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/654375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/75/654375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/75/654375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/75/654375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/75/654375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/75/654375/1"}}},"commit":{"parents":[{"commit":"70afbfc3ce7eafb94bff6b537cda81ab28279fc5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/70afbfc3ce7eafb94bff6b537cda81ab28279fc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:15:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:15:55.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie11d5dcf5912f23684ebd93ff0548791d57bded6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
