)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~655309","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Iea4a1b93fd71714e2dfd0d55c87d7dae6bb1797d","project":"openstack/openstack-ansible-os_rally","branch":"master","hashtags":[],"change_id":"Iea4a1b93fd71714e2dfd0d55c87d7dae6bb1797d","subject":"update git.openstack.org to opendev","status":"ABANDONED","created":"2019-04-24 06:29:16.000000000","updated":"2019-07-19 17:27:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"522b1633253719c2b572f9e243db5e26d625d0a9","_number":655309,"virtual_id_number":655309,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-04-24 14:55:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"all":[{"value":1,"date":"2019-07-14 15:21:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 07:41:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 14:55:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-14 15:21:51.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"}],"messages":[{"id":"3d908c00699c430523e5e1faeae6522f56ad21dd","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 06:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1674390bc460a86e0e9aae3612ec33d02b1a7671","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-24 07:41:16.000000000","message":"Patch Set 1: Code-Review-1\n\nSee comments in https://review.opendev.org/#/c/654744","accounts_in_message":[],"_revision_number":1},{"id":"1df3dfd0e1db0f03b9406b36fe2ee8be4de23e88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:58:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/09/655309/1/check/openstack-ansible-linters/cf488af/ : FAILURE in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/09/655309/1/check/openstack-tox-docs/4295c9b/html/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/09/655309/1/check/build-openstack-releasenotes/9bf5bb8/html/ : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/09/655309/1/check/openstack-ansible-deploy-aio_metal_rally-centos-7/b80a355/ : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/09/655309/1/check/openstack-ansible-deploy-aio_metal_rally-debian-stable/f49cb6a/ : SUCCESS in 59m 43s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/09/655309/1/check/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/1b51d94/ : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"20643aa6ce8e50c30be7011f666f1666b386ff08","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 08:17:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e273da970b5836c0fde3677b5478cfa90bdfe7c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:42:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/09/655309/2/check/openstack-ansible-linters/ded5bf6/ : FAILURE in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/09/655309/2/check/openstack-tox-docs/816da1a/html/ : SUCCESS in 3m 59s\n- build-openstack-releasenotes http://logs.openstack.org/09/655309/2/check/build-openstack-releasenotes/1648765/html/ : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/09/655309/2/check/openstack-ansible-deploy-aio_metal_rally-centos-7/598d2a9/ : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/09/655309/2/check/openstack-ansible-deploy-aio_metal_rally-debian-stable/014bbeb/ : SUCCESS in 52m 12s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/09/655309/2/check/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/ac091db/ : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":2},{"id":"69e9aaafe52fba648486d7069708a8d1ef1e8621","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 14:45:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a8220eeddeae6e4876c68ad03b41e37efae175a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:55:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/655309/3/check/openstack-ansible-linters/c234e02/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/09/655309/3/check/openstack-tox-docs/57e78a5/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/09/655309/3/check/build-openstack-releasenotes/bc848b5/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"91bcb6acc2e1f541ada93cc5ffecda11bfe6215a","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:21:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"522b1633253719c2b572f9e243db5e26d625d0a9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 17:27:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a19c0c51b02b6af36261ea283b0dbf1dac376f70","revisions":{"54e86b106a7935c34942dfd9715d14550c29c1fc":{"kind":"REWORK","_number":1,"created":"2019-04-24 06:29:16.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/09/655309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/09/655309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/1"}}},"commit":{"parents":[{"commit":"b7cdb5c3ea34441015275915541a7ec671f9cd2c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:28:28.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:29:00.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Iea4a1b93fd71714e2dfd0d55c87d7dae6bb1797d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/54e86b106a7935c34942dfd9715d14550c29c1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/54e86b106a7935c34942dfd9715d14550c29c1fc"}]},"branch":"refs/heads/master"},"96457ca940c2ab827f4ce8a2ff3629573b03b81e":{"kind":"REWORK","_number":2,"created":"2019-04-24 08:17:03.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/09/655309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/09/655309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/2"}}},"commit":{"parents":[{"commit":"b7cdb5c3ea34441015275915541a7ec671f9cd2c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:28:28.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 08:16:55.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Iea4a1b93fd71714e2dfd0d55c87d7dae6bb1797d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/96457ca940c2ab827f4ce8a2ff3629573b03b81e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/96457ca940c2ab827f4ce8a2ff3629573b03b81e"}]},"branch":"refs/heads/master"},"a19c0c51b02b6af36261ea283b0dbf1dac376f70":{"kind":"REWORK","_number":3,"created":"2019-04-24 14:45:33.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/09/655309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/09/655309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/09/655309/3"}}},"commit":{"parents":[{"commit":"b7cdb5c3ea34441015275915541a7ec671f9cd2c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:28:28.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 14:45:22.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Iea4a1b93fd71714e2dfd0d55c87d7dae6bb1797d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/a19c0c51b02b6af36261ea283b0dbf1dac376f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/a19c0c51b02b6af36261ea283b0dbf1dac376f70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
