)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~658037","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Id5637db09cc9317b1632b6057415705a2431bd57","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id5637db09cc9317b1632b6057415705a2431bd57","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:35:26.000000000","updated":"2019-05-09 15:16:03.000000000","submitted":"2019-05-09 15:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658037-1557414963301-4a183efd","meta_rev_id":"aa45bd2686b8abe5e8660314eb9407f4bf970c10","_number":658037,"virtual_id_number":658037,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 15:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:56:57.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 15:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81e5154c1c268cf21cf6864a411441118658373f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50b9cbcbe34c3ef836e749a964c4ebf31830ba50","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:56:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f216e092ac01640779cdcc242539ad8d2a82e04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/658037/1/check/openstack-ansible-linters/157dfde/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/37/658037/1/check/openstack-tox-docs/37cd55b/html/ : SUCCESS in 2m 57s\n- build-openstack-releasenotes http://logs.openstack.org/37/658037/1/check/build-openstack-releasenotes/e7b5df2/html/ : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/37/658037/1/check/openstack-ansible-deploy-aio_metal_rally-centos-7/6dd44d2/ : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/37/658037/1/check/openstack-ansible-deploy-aio_metal_rally-debian-stable/0c2d75a/ : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/37/658037/1/check/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/72931a9/ : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f44796064ce2720788788bd1d39686fe410b9db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:17:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0188f7e542f31eef040e7aaeec46a8a90fd5ff42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:16:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0295b67f5075e2f2e834422a2c094fcf4bfa6ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:16:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/658037/1/gate/openstack-ansible-linters/cc663d8/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/37/658037/1/gate/openstack-tox-docs/a606f67/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/37/658037/1/gate/build-openstack-releasenotes/d38a283/html/ : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_metal_rally-centos-7 http://logs.openstack.org/37/658037/1/gate/openstack-ansible-deploy-aio_metal_rally-centos-7/4894c39/ : SUCCESS in 1h 56m 59s\n- openstack-ansible-deploy-aio_metal_rally-debian-stable http://logs.openstack.org/37/658037/1/gate/openstack-ansible-deploy-aio_metal_rally-debian-stable/870aa51/ : SUCCESS in 55m 30s\n- openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic http://logs.openstack.org/37/658037/1/gate/openstack-ansible-deploy-aio_metal_rally-ubuntu-bionic/96799dd/ : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e18d963662d2c10e495f30232bd1b7affd42eb88","revisions":{"e18d963662d2c10e495f30232bd1b7affd42eb88":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:35:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/658037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/37/658037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/37/658037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/37/658037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/37/658037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/37/658037/1"}}},"commit":{"parents":[{"commit":"b7cdb5c3ea34441015275915541a7ec671f9cd2c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b7cdb5c3ea34441015275915541a7ec671f9cd2c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:25.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id5637db09cc9317b1632b6057415705a2431bd57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e18d963662d2c10e495f30232bd1b7affd42eb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e18d963662d2c10e495f30232bd1b7affd42eb88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
