)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~660284","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Ie2246172ba6e9cfb0a25efdb6ba3d8b2909c8705","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"Ie2246172ba6e9cfb0a25efdb6ba3d8b2909c8705","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-21 08:29:41.000000000","updated":"2019-06-03 17:08:25.000000000","submitted":"2019-06-03 17:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660284-1559581705578-02bd17ee","meta_rev_id":"ca4c91a614558b0bafdf6c244c05bc4c55df5d2f","_number":660284,"virtual_id_number":660284,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-03 17:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-22 14:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-22 14:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-22 14:32:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-06-03 15:09:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-27 19:52:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 14:32:23.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-27 19:52:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-03 15:09:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-03 17:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"819eafe26a3d046eb742ae42fd2db41f4c9374f3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f846e947ac5ca3ec93ea87219f8a98c9c4eed463","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:50:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/660284/1/check/openstack-ansible-linters/21f450f/ : SUCCESS in 5m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660284/1/check/openstack-ansible-deploy-aio_metal-debian-stable/44ee872/ : SUCCESS in 59m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660284/1/check/openstack-ansible-deploy-aio_metal-centos-7/0d502f5/ : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660284/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/07fe459/ : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660284/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8635c51/ : SUCCESS in 1h 04m 32s\n- openstack-tox-docs http://logs.openstack.org/84/660284/1/check/openstack-tox-docs/d26a0c7/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9c4571efda4fa21d1e2d19bae3ec2f9972b61bf4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:37:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"771f90fe601e6f8d548fadf2be679425763dd8b6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:38:54.000000000","message":"Removed Code-Review+2 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"51777b7136b033711459a539e2ebb1aa4aa89554","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 15:18:41.000000000","message":"Patch Set 1: Code-Review-1\n\nRally role is not executing in the jobs","accounts_in_message":[],"_revision_number":1},{"id":"c35266437a1781b3458d10f8d844f3599caa9413","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 07:48:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d93afd9cb3f3c031da46783d76a55dfe30e6d325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:10:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/84/660284/2/check/openstack-ansible-linters/108da77/ : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660284/2/check/openstack-ansible-deploy-aio_metal-debian-stable/4c1857d/ : FAILURE in 52m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660284/2/check/openstack-ansible-deploy-aio_metal-centos-7/d720f37/ : FAILURE in 1h 15m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660284/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/3ebb473/ : FAILURE in 1h 06m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660284/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/21a5404/ : FAILURE in 1h 03m 14s\n- openstack-tox-docs http://logs.openstack.org/84/660284/2/check/openstack-tox-docs/f4d806c/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1b7bfda5bbfad63e227c80a9aaa07d16b93fb4a7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 12:34:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4def990f39d07afdb2a1f0b46effac37bc424dd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 14:04:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/660284/3/check/openstack-ansible-linters/d132ade/ : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660284/3/check/openstack-ansible-deploy-aio_metal-debian-stable/942d05f/ : SUCCESS in 1h 03m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660284/3/check/openstack-ansible-deploy-aio_metal-centos-7/c393189/ : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660284/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/6c7cc29/ : SUCCESS in 1h 07m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660284/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/08101cd/ : SUCCESS in 54m 48s\n- openstack-tox-docs http://logs.openstack.org/84/660284/3/check/openstack-tox-docs/9cfdc8a/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c17f99ed515f0ac6e83e4d05be184a69b0939499","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-22 14:21:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"72f9f94ac99c3a0c10be3dd1eb812cde2539df8c","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 14:32:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c41034366122e2f6944cc6322995f143adc0637","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-27 19:52:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dfe400e63f6adc61a42d4cb46078c4d79e485059","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-27 19:58:08.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"cbb6a94f40745c85deab4b624e3a8487dbcae37a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-03 15:09:25.000000000","message":"Patch Set 3: Workflow+1\n\npush to gate","accounts_in_message":[],"_revision_number":3},{"id":"53324628c6d931fccf3c65d7f6b78b92b4096af9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:09:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fcfef9360232c76fccb34ca58aa9d73d6f014105","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:08:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"aa626c9824abaef16b059d64a0237c4f19b3d2a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:08:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/660284/3/gate/openstack-ansible-linters/c996146/ : SUCCESS in 4m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660284/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/de21b85/ : SUCCESS in 59m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660284/3/gate/openstack-ansible-deploy-aio_metal-centos-7/fb85e8e/ : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660284/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/0689e73/ : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660284/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a2b379c/ : SUCCESS in 51m 20s\n- openstack-tox-docs http://logs.openstack.org/84/660284/3/gate/openstack-tox-docs/805b051/html/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b94df04f449056c411a042c80a93461477630b96","revisions":{"668847330949ef20c0638f9ccc724fee433ed48c":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:29:41.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/84/660284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/84/660284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/1"}}},"commit":{"parents":[{"commit":"e18d963662d2c10e495f30232bd1b7affd42eb88","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e18d963662d2c10e495f30232bd1b7affd42eb88"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:28:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:29:22.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nChange-Id: Ie2246172ba6e9cfb0a25efdb6ba3d8b2909c8705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/668847330949ef20c0638f9ccc724fee433ed48c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/668847330949ef20c0638f9ccc724fee433ed48c"}]},"branch":"refs/heads/master"},"608817d1545c0015c33477da0f9eead479cbae2f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-22 07:48:35.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/84/660284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/84/660284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/2"}}},"commit":{"parents":[{"commit":"e18d963662d2c10e495f30232bd1b7affd42eb88","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e18d963662d2c10e495f30232bd1b7affd42eb88"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:28:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-22 07:48:33.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nDepends-On: https://review.opendev.org/660591\n\nChange-Id: Ie2246172ba6e9cfb0a25efdb6ba3d8b2909c8705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/608817d1545c0015c33477da0f9eead479cbae2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/608817d1545c0015c33477da0f9eead479cbae2f"}]},"branch":"refs/heads/master"},"b94df04f449056c411a042c80a93461477630b96":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-22 12:34:28.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/84/660284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/84/660284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/84/660284/3"}}},"commit":{"parents":[{"commit":"dbc71ac98cc2f1abd72e7bfc18443847e1ae028a","subject":"Bump rally versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/dbc71ac98cc2f1abd72e7bfc18443847e1ae028a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:28:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-22 12:34:28.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nDepends-On: https://review.opendev.org/660591\n\nChange-Id: Ie2246172ba6e9cfb0a25efdb6ba3d8b2909c8705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b94df04f449056c411a042c80a93461477630b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b94df04f449056c411a042c80a93461477630b96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
