)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~660286","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I4d86070e9e52ed97ce890c3366967263559c04d7","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I4d86070e9e52ed97ce890c3366967263559c04d7","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-21 08:32:11.000000000","updated":"2019-06-05 13:15:48.000000000","submitted":"2019-06-05 13:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660286-1559740548368-2368eebd","meta_rev_id":"2eee7d255f05d721630330cb5d11d02be2585627","_number":660286,"virtual_id_number":660286,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-05 13:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":1,"date":"2019-06-05 07:43:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-05 07:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-05 07:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-05 07:43:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 10:10:08.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-21 11:05:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-27 07:14:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-05 07:43:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-05 13:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"747de4c85e6c83d6eed034c5c205cb2c00bfc25c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:32:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5b5bf4e396d604c1513a7c38d176ad363905e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:02:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/86/660286/1/check/openstack-ansible-linters/cc121fd/ : SUCCESS in 5m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/86/660286/1/check/openstack-ansible-deploy-aio_metal-debian-stable/b82a03e/ : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/86/660286/1/check/openstack-ansible-deploy-aio_metal-centos-7/8c3d892/ : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/86/660286/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/6d5bb2a/ : SUCCESS in 58m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/86/660286/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ee0c40f/ : SUCCESS in 59m 13s\n- openstack-tox-docs http://logs.openstack.org/86/660286/1/check/openstack-tox-docs/51170b9/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5e98e484733c9fcfc95e2787249ee751a85a0107","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-21 10:10:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10d86fab9f873280b1e7c96dcdb6cb999348728e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-21 10:41:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ce9afae2ff6756200df9431aad6a2d4009103ca","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:05:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f353050bf18e69836975b166057469bcab10f09c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 07:14:03.000000000","message":"Patch Set 1: Workflow+1\n\nkick into gate","accounts_in_message":[],"_revision_number":1},{"id":"a0bf972c04122fea5d0552429104e87fe99f7f1d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-03 18:52:02.000000000","message":"Patch Set 1: Workflow+1\n\nanother kick into gate","accounts_in_message":[],"_revision_number":1},{"id":"570dbf1c054775652bc9e160e5b120e3359c1d5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 07:42:12.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"9117b17178f2a62a2b85fe5773a8d56f9408f0b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 07:42:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e9577da39102904b50c3100160b7d0ee09a9771","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 07:43:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ec79ff45f11acd27d101e9c4e62bbbf44adcfaec","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 07:43:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"befaffe3d5aeb3f73ca4953f13e636efa2d8868e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 09:30:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/86/660286/2/check/openstack-ansible-linters/eb55f88/ : FAILURE in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-debian-stable/e8f3b6c/ : SUCCESS in 52m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-centos-7/c2d4475/ : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/c3490c4/ : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1f0ab35/ : SUCCESS in 1h 13m 14s\n- openstack-tox-docs http://logs.openstack.org/86/660286/2/check/openstack-tox-docs/adcbee7/html/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a1b840bfb9022b9d81773667cce6e022fda210f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 09:52:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"490324566f791e4c5c3342127c022a78c32bce78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 11:43:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/86/660286/2/check/openstack-ansible-linters/f524cb0/ : SUCCESS in 8m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-debian-stable/28cb3a9/ : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-centos-7/663499e/ : SUCCESS in 1h 22m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/fa6cee2/ : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/86/660286/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/82ef0ab/ : SUCCESS in 55m 37s\n- openstack-tox-docs http://logs.openstack.org/86/660286/2/check/openstack-tox-docs/8bb169c/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"49fd2cf825843f0699c8eb4fb576b865d8168e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 11:43:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52c81442abc1001192c17166c7249d5ecd6d0544","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 13:15:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/86/660286/2/gate/openstack-ansible-linters/0071751/ : SUCCESS in 4m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/86/660286/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/4d80ca2/ : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/86/660286/2/gate/openstack-ansible-deploy-aio_metal-centos-7/68bd3a5/ : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/86/660286/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/1aa910d/ : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/86/660286/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d44e139/ : SUCCESS in 1h 01m 41s\n- openstack-tox-docs http://logs.openstack.org/86/660286/2/gate/openstack-tox-docs/3d34a9b/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"de01a289aafe0a4ee8c8ab420fd12d72c093eb32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 13:15:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83b5bf4e56acfca7c0b6de17debf96bbdb17bb05","revisions":{"d28a39826ef5f6106ddc5c2915cf22ae93da93d0":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:32:11.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/86/660286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/86/660286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/1"}}},"commit":{"parents":[{"commit":"668847330949ef20c0638f9ccc724fee433ed48c","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/668847330949ef20c0638f9ccc724fee433ed48c"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:31:21.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:31:21.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I4d86070e9e52ed97ce890c3366967263559c04d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/d28a39826ef5f6106ddc5c2915cf22ae93da93d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/d28a39826ef5f6106ddc5c2915cf22ae93da93d0"}]},"branch":"refs/heads/master"},"83b5bf4e56acfca7c0b6de17debf96bbdb17bb05":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-05 07:43:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/660286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/86/660286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/86/660286/2"}}},"commit":{"parents":[{"commit":"b94df04f449056c411a042c80a93461477630b96","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b94df04f449056c411a042c80a93461477630b96"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:31:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-06-05 07:43:18.000000000","tz":0},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I4d86070e9e52ed97ce890c3366967263559c04d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/83b5bf4e56acfca7c0b6de17debf96bbdb17bb05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/83b5bf4e56acfca7c0b6de17debf96bbdb17bb05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
