)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~664340","triplet_id":"openstack%2Fopenstack-ansible-os_rally~stable%2Frocky~Ib872228e1f9414cb6493cb0044a404a09fc859cd","project":"openstack/openstack-ansible-os_rally","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ib872228e1f9414cb6493cb0044a404a09fc859cd","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-06-10 15:45:03.000000000","updated":"2021-01-13 18:00:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9fa4352c3f81c0024a2ba2bf777d39c6cd0b40be","_number":664340,"virtual_id_number":664340,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-17 14:00:47.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":-1,"date":"2019-06-10 16:57:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-15 19:04:28.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 16:57:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 14:00:47.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-07-15 19:04:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0c1f2bc93b2f8ef32990e7b2287cd304faca59e4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1d9300e4a5ad27bd7b9e4ea413602b8ae3f961b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:57:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664340/1/check/openstack-ansible-linters/33097d8/ : SUCCESS in 9m 35s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-centos-7/f0a230c/ : FAILURE in 31m 00s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-423/675038c/ : FAILURE in 7m 13s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-150/ccec262/ : FAILURE in 14m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-bionic/aa71ef6/ : FAILURE in 39m 44s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-xenial/40eb133/ : FAILURE in 34m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/664340/1/check/build-openstack-sphinx-docs/6969d17/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c35f914982bb100a3f1c4d9a28289cc6607e6726","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:41:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27ac8462890ab90abe612e19a6cc8e415df92e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:13:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664340/1/check/openstack-ansible-linters/daef1c1/ : SUCCESS in 5m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-centos-7/52d881b/ : FAILURE in 29m 49s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-423/1593454/ : FAILURE in 7m 38s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-150/9517d64/ : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-bionic/756f2fd/ : FAILURE in 23m 21s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-xenial/8e43c0f/ : FAILURE in 22m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/664340/1/check/build-openstack-sphinx-docs/041a256/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b0e90527db56b586b25d0cb5c3fcb6e4e7296782","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-12 12:21:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"323c4bc133fce6db8a291fe284a1aac2cc74f027","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 13:03:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664340/1/check/openstack-ansible-linters/753f600/ : SUCCESS in 4m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-centos-7/6633dec/ : FAILURE in 30m 40s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-423/c597ad2/ : FAILURE in 8m 25s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-150/ae061f4/ : FAILURE in 8m 54s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-bionic/0e19926/ : FAILURE in 25m 50s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-xenial/2154922/ : FAILURE in 20m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/664340/1/check/build-openstack-sphinx-docs/1bb1f4c/html/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ab9a7f4780cc92532317f1c192d32568434ae278","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-17 14:00:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cebc772fd89196563c6923f63bef569e856a667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664340/1/check/openstack-ansible-linters/50daf35/ : SUCCESS in 5m 57s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-centos-7/506c181/ : FAILURE in 33m 37s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-423/7a75386/ : FAILURE in 16m 55s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-150/d3d58d1/ : FAILURE in 10m 18s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-bionic/87ec659/ : FAILURE in 23m 13s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-xenial/bdeb305/ : FAILURE in 30m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/664340/1/check/build-openstack-sphinx-docs/52b66dc/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2995b577671a4ae89ad33f3602d164229b248359","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 19:04:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3db3f22c981f22d49f0acbc57c22c4dae8f981b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:50:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664340/1/check/openstack-ansible-linters/7dea513/ : SUCCESS in 5m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-centos-7/6530832/ : FAILURE in 30m 39s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-423/7b175e6/ : FAILURE in 8m 09s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-opensuse-150/671af9f/ : FAILURE in 12m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-bionic/271f3ee/ : FAILURE in 26m 52s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/40/664340/1/check/openstack-ansible-functional-ubuntu-xenial/ded140f/ : FAILURE in 38m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/664340/1/check/build-openstack-sphinx-docs/36fc5e9/html/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9fa4352c3f81c0024a2ba2bf777d39c6cd0b40be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-13 18:00:17.000000000","message":"Abandoned\n\nrocky is extended-maintainance, abandoning","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2912023eb225aff658879a230ad03912673411bc","revisions":{"2912023eb225aff658879a230ad03912673411bc":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/664340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/40/664340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/40/664340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/40/664340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/40/664340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/40/664340/1"}}},"commit":{"parents":[{"commit":"3e8a89f72c8248895bdb1080bdd3de928fa1b38c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/3e8a89f72c8248895bdb1080bdd3de928fa1b38c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:02.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ib872228e1f9414cb6493cb0044a404a09fc859cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2912023eb225aff658879a230ad03912673411bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2912023eb225aff658879a230ad03912673411bc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
