)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~669694","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I2b09e3cde1d2dd51f2d352d4eb57a90751fe854f","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"deprecation-warning","hashtags":[],"change_id":"I2b09e3cde1d2dd51f2d352d4eb57a90751fe854f","subject":"Fix deprecation warning","status":"MERGED","created":"2019-07-08 14:59:39.000000000","updated":"2019-07-10 17:45:04.000000000","submitted":"2019-07-10 17:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669694-1562780704091-b2961d89","meta_rev_id":"f903284d87f87f22ccdfff1db2d3cacfda1b55bc","_number":669694,"virtual_id_number":669694,"owner":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 17:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-10 16:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-10 10:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-10 16:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 10:35:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-10 16:01:05.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 17:45:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfbffcba3b724bb11f3eb13b108f793ecce35454","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-08 14:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6b672cfc6956c5477110742d459758ccc52168b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:33:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/94/669694/1/check/openstack-ansible-linters/c889870/ : SUCCESS in 5m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/94/669694/1/check/openstack-ansible-deploy-aio_metal-debian-stable/d383f87/ : SUCCESS in 1h 02m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/94/669694/1/check/openstack-ansible-deploy-aio_metal-centos-7/bb94c68/ : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/94/669694/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/f0a3f63/ : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/94/669694/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3d8079d/ : SUCCESS in 1h 10m 04s\n- openstack-tox-docs http://logs.openstack.org/94/669694/1/check/openstack-tox-docs/ac7b87f/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6faaf5b26616d3ec106a14c0a32c597e3fc21d7b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 10:35:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"baf95151bb36de7f43908b474383f965c410612b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 16:01:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"148c5441d1a942e7c1df35f2670aef1cb684c2be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:01:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85c11d8c7f9892275d8bd89def7c59fdb1b88534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:45:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/94/669694/1/gate/openstack-ansible-linters/a9a3c22/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/94/669694/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/ac0a583/ : SUCCESS in 1h 05m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/94/669694/1/gate/openstack-ansible-deploy-aio_metal-centos-7/b31a472/ : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/94/669694/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/f6b6555/ : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/94/669694/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/635f3d5/ : SUCCESS in 59m 48s\n- openstack-tox-docs http://logs.openstack.org/94/669694/1/gate/openstack-tox-docs/cb1de18/html/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8395a1bcd12242581cd5e80b2d597527ab60bb70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:45:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1bfdcec891a679b4e7f7c241d0da7c361564d57","revisions":{"e1bfdcec891a679b4e7f7c241d0da7c361564d57":{"kind":"REWORK","_number":1,"created":"2019-07-08 14:59:39.000000000","uploader":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"ref":"refs/changes/94/669694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/94/669694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/94/669694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/94/669694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/94/669694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/94/669694/1"}}},"commit":{"parents":[{"commit":"15c7e61436d5ed2fc11a781ccdfe1dcf0a163981","subject":"Merge \"db_setup: refactor database setup to a common file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/15c7e61436d5ed2fc11a781ccdfe1dcf0a163981"}]}],"author":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:57:59.000000000","tz":120},"committer":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-08 14:59:55.000000000","tz":120},"subject":"Fix deprecation warning","message":"Fix deprecation warning\n\nTASK [os_rally : Create/upgrade Rally DB schema] ***********\n[DEPRECATION WARNING]: Using tests as filters is deprecated.\nInstead of using `result|search` use `result is search`.\nThis feature will be removed in version 2.9.\nDeprecation warnings can be disabled by setting\ndeprecation_warnings\u003dFalse in ansible.cfg.\n\nChange-Id: I2b09e3cde1d2dd51f2d352d4eb57a90751fe854f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e1bfdcec891a679b4e7f7c241d0da7c361564d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/e1bfdcec891a679b4e7f7c241d0da7c361564d57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
