)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~670054","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I73a8bb8682d3daf027bc3b532912c852121afbae","project":"openstack/openstack-ansible-os_rally","branch":"master","hashtags":[],"change_id":"I73a8bb8682d3daf027bc3b532912c852121afbae","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-07-10 13:13:12.000000000","updated":"2019-07-16 17:34:48.000000000","submitted":"2019-07-16 17:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670054-1563298488924-e47bf6dd","meta_rev_id":"7d2001b53fca81239b1457795f6935b00a6c1e5a","_number":670054,"virtual_id_number":670054,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-16 17:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"all":[{"value":1,"date":"2019-07-13 19:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":1,"date":"2019-07-14 02:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-07-14 02:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-16 14:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 15:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-16 15:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-13 19:33:51.000000000","updated_by":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"reviewer":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"state":"REVIEWER"},{"updated":"2019-07-14 02:31:29.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 02:46:54.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-16 14:29:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-16 15:56:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-16 17:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d92cde8f29ee59a20a02860afa195479283e7c4b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 13:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78bbc2a23b92140f2520cef53c049b0447b5b355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 14:41:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/670054/1/check/openstack-ansible-linters/dcb4961/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/670054/1/check/openstack-ansible-deploy-aio_metal-debian-stable/883bfc1/ : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/670054/1/check/openstack-ansible-deploy-aio_metal-centos-7/19935d6/ : SUCCESS in 1h 23m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/54/670054/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/123dcf9/ : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/670054/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c070697/ : SUCCESS in 57m 44s\n- openstack-tox-docs http://logs.openstack.org/54/670054/1/check/openstack-tox-docs/c044b6c/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes http://logs.openstack.org/54/670054/1/check/build-openstack-releasenotes/65001f2/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"074d3d5c17bbdd1fb8a3b220288bffc3edb20572","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 15:05:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f288fb30d268f280a6d74d77ef92be8a7af739da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:41:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/670054/2/check/openstack-ansible-linters/0fa3006/ : SUCCESS in 9m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/670054/2/check/openstack-ansible-deploy-aio_metal-debian-stable/daa85c5/ : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/670054/2/check/openstack-ansible-deploy-aio_metal-centos-7/bff4ba5/ : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/54/670054/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/9dc030f/ : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/670054/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e3e8df0/ : SUCCESS in 1h 20m 29s\n- openstack-tox-docs http://logs.openstack.org/54/670054/2/check/openstack-tox-docs/abb34c4/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/54/670054/2/check/build-openstack-releasenotes/e8608c7/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"aee5c5171e75cc2ec202518b4c464a6068dea171","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 22:05:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"71cf170f3f3385f6a5224bf8bbd706b13857c51c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 23:37:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/670054/3/check/openstack-ansible-linters/78fb363/ : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/670054/3/check/openstack-ansible-deploy-aio_metal-debian-stable/7c82737/ : SUCCESS in 57m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/670054/3/check/openstack-ansible-deploy-aio_metal-centos-7/c6f98aa/ : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/54/670054/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/14889c4/ : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/670054/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9702d76/ : SUCCESS in 58m 17s\n- openstack-tox-docs http://logs.openstack.org/54/670054/3/check/openstack-tox-docs/869e9d1/html/ : SUCCESS in 3m 27s\n- build-openstack-releasenotes http://logs.openstack.org/54/670054/3/check/build-openstack-releasenotes/f3b4642/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f35cdf5c98e70bdeacea657871e8af6a82820239","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 19:33:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"23beeb5620ab5b997c09f1a279c72117af097eaf","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-14 02:31:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2d4a6e544fa74a84a11a16a6880f5882b11d6f9c","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 02:46:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"65e13b9d24660c394dc4349ec5096d3c6f994e00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 14:29:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf888359a7ff95ce774017ca175ec94711412e08","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 15:56:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a4704a2bd60a519a80041c0a021e460c4b51f3d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:56:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c5bdcd2fc5bea924726404db18b07c7f747aed39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:34:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/670054/3/gate/openstack-ansible-linters/dae8df3/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/670054/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/d4b31e0/ : SUCCESS in 1h 04m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/670054/3/gate/openstack-ansible-deploy-aio_metal-centos-7/dd2d148/ : SUCCESS in 1h 37m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/54/670054/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/8bcd8da/ : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/670054/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4bd368b/ : SUCCESS in 1h 01m 01s\n- openstack-tox-docs http://logs.openstack.org/54/670054/3/gate/openstack-tox-docs/bd684cb/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/54/670054/3/gate/build-openstack-releasenotes/38c45b8/html/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7d2001b53fca81239b1457795f6935b00a6c1e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c53e631763c1c51e6e3613367f47d50696c204c9","revisions":{"7bebcbcfe3603c8604cbc09d28cfd3cb52cb0d97":{"kind":"REWORK","_number":1,"created":"2019-07-10 13:13:12.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/54/670054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/54/670054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/1"}}},"commit":{"parents":[{"commit":"15c7e61436d5ed2fc11a781ccdfe1dcf0a163981","subject":"Merge \"db_setup: refactor database setup to a common file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/15c7e61436d5ed2fc11a781ccdfe1dcf0a163981"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 13:12:18.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 13:13:04.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I73a8bb8682d3daf027bc3b532912c852121afbae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/7bebcbcfe3603c8604cbc09d28cfd3cb52cb0d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/7bebcbcfe3603c8604cbc09d28cfd3cb52cb0d97"}]},"branch":"refs/heads/master"},"85f526db84beb283dab89ee101dc9cd684b18b30":{"kind":"REWORK","_number":2,"created":"2019-07-10 15:05:49.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/54/670054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/54/670054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/2"}}},"commit":{"parents":[{"commit":"15c7e61436d5ed2fc11a781ccdfe1dcf0a163981","subject":"Merge \"db_setup: refactor database setup to a common file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/15c7e61436d5ed2fc11a781ccdfe1dcf0a163981"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 13:12:18.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 15:05:08.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I73a8bb8682d3daf027bc3b532912c852121afbae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/85f526db84beb283dab89ee101dc9cd684b18b30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/85f526db84beb283dab89ee101dc9cd684b18b30"}]},"branch":"refs/heads/master"},"c53e631763c1c51e6e3613367f47d50696c204c9":{"kind":"REWORK","_number":3,"created":"2019-07-10 22:05:10.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/54/670054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/54/670054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/54/670054/3"}}},"commit":{"parents":[{"commit":"15c7e61436d5ed2fc11a781ccdfe1dcf0a163981","subject":"Merge \"db_setup: refactor database setup to a common file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/15c7e61436d5ed2fc11a781ccdfe1dcf0a163981"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 13:12:18.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 22:04:51.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I73a8bb8682d3daf027bc3b532912c852121afbae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/c53e631763c1c51e6e3613367f47d50696c204c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/c53e631763c1c51e6e3613367f47d50696c204c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
