)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~671569","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Icfac74a031d9059d50772d8423c5ba3f6aa00407","project":"openstack/openstack-ansible-os_rally","branch":"master","hashtags":[],"change_id":"Icfac74a031d9059d50772d8423c5ba3f6aa00407","subject":"build: use upper-constraints.txt from rally-openstack","status":"MERGED","created":"2019-07-18 20:43:36.000000000","updated":"2019-07-19 14:02:36.000000000","submitted":"2019-07-19 14:02:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671569-1563544956774-5870c8eb","meta_rev_id":"2ade7fc84d4f924c976a69ceba43cc7197f3a486","_number":671569,"virtual_id_number":671569,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-19 14:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-07-19 12:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 12:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-07-19 12:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:26:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 12:30:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-19 14:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"502a348071bb45cd18953f38a6e27d52b168e411","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 20:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94d29005e42292b64ff90a89605140ab69a468dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:30:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/671569/1/check/openstack-ansible-linters/d78724f/ : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/69/671569/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0e70685/ : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/69/671569/1/check/openstack-ansible-deploy-aio_metal-centos-7/6dc70b5/ : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/69/671569/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/48db325/ : SUCCESS in 1h 44m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/69/671569/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9b8ff7a/ : SUCCESS in 1h 21m 54s\n- openstack-tox-docs http://logs.openstack.org/69/671569/1/check/openstack-tox-docs/28aeac2/html/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9ab8c5f0da47e9fbb383aa077a147e9791592a3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 12:26:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e890fe83f5ca3116cdc9d5942c871e83aefda8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50270872332f704356eaddefe5767a47576a4be9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:30:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ffe2043f0dde46879161231b2369f3f57dc20b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:02:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/671569/1/gate/openstack-ansible-linters/23b9c5e/ : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/69/671569/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/594da09/ : SUCCESS in 53m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/69/671569/1/gate/openstack-ansible-deploy-aio_metal-centos-7/c727f2d/ : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/69/671569/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/91fe7f7/ : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/69/671569/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3d61463/ : SUCCESS in 1h 03m 36s\n- openstack-tox-docs http://logs.openstack.org/69/671569/1/gate/openstack-tox-docs/1d24b8f/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2ade7fc84d4f924c976a69ceba43cc7197f3a486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:02:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cce89200844d70efb23685dd3e8df08c9f58165","revisions":{"1cce89200844d70efb23685dd3e8df08c9f58165":{"kind":"REWORK","_number":1,"created":"2019-07-18 20:43:36.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/69/671569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/69/671569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/69/671569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/69/671569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/69/671569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/69/671569/1"}}},"commit":{"parents":[{"commit":"bf1f02dbcb5b5be0e83b41beb7203df03f23610b","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/bf1f02dbcb5b5be0e83b41beb7203df03f23610b"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-18 20:42:53.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-18 20:43:34.000000000","tz":-240},"subject":"build: use upper-constraints.txt from rally-openstack","message":"build: use upper-constraints.txt from rally-openstack\n\nWe had all of the upper-constraints hardwired inside the role\nwhich was not ideal.  This patch grabs the latest upper-constraints\nas we are installing the latest package in this case.\n\nChange-Id: Icfac74a031d9059d50772d8423c5ba3f6aa00407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/1cce89200844d70efb23685dd3e8df08c9f58165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/1cce89200844d70efb23685dd3e8df08c9f58165"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
