)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~671824","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I3c4b2fab82eb6047b4ce37119c1af18120234e43","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I3c4b2fab82eb6047b4ce37119c1af18120234e43","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-19 18:25:45.000000000","updated":"2019-07-20 16:34:55.000000000","submitted":"2019-07-20 16:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671824-1563640495774-4e3e88c1","meta_rev_id":"2bc0d03d8b9f29f720303d24e95f90fde3284220","_number":671824,"virtual_id_number":671824,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-20 16:34:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-20 16:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-20 16:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-20 16:24:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-20 16:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e882b8e37109da1826fe39dd34dacaebedd4b5e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38edeb4bef05828491efa2f461c6b6b33e9df33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:53:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/24/671824/1/check/requirements-check/a3e9ee8/ : SUCCESS in 3m 59s\n- openstack-ansible-linters http://logs.openstack.org/24/671824/1/check/openstack-ansible-linters/da4a6b4/ : SUCCESS in 9m 49s\n- openstack-tox-docs http://logs.openstack.org/24/671824/1/check/openstack-tox-docs/73fa20f/ : POST_FAILURE in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/24/671824/1/check/build-openstack-releasenotes/74ef4a9/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bc16b6a44f309f538d9a163c17b8ff09a3be42cd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:37:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46f2ea67b59acb6eeeba5da02303d7db0dbc6175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:50:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/24/671824/2/check/requirements-check/adcc170/ : SUCCESS in 3m 14s\n- openstack-ansible-linters http://logs.openstack.org/24/671824/2/check/openstack-ansible-linters/4651a93/ : SUCCESS in 9m 52s\n- openstack-tox-docs http://logs.openstack.org/24/671824/2/check/openstack-tox-docs/f779451/html/ : SUCCESS in 3m 38s\n- build-openstack-releasenotes http://logs.openstack.org/24/671824/2/check/build-openstack-releasenotes/aa2562a/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0956e59df06d16745d845dc96b54f2d3d4ce5c57","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-20 16:24:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e7bf694ff39299e902c176c0ece32e498daca63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:25:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c6bec28e19a4863f9b903f27a674a966de2cf61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:34:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"73c96662fe2987324fa226d485289b8df32bdd3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:34:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/24/671824/2/gate/requirements-check/739b6a1/ : SUCCESS in 1m 54s\n- openstack-ansible-linters http://logs.openstack.org/24/671824/2/gate/openstack-ansible-linters/8c02cb1/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/24/671824/2/gate/openstack-tox-docs/cb9cbc1/html/ : SUCCESS in 2m 22s\n- build-openstack-releasenotes http://logs.openstack.org/24/671824/2/gate/build-openstack-releasenotes/b91368c/html/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5ceb57e33a7d52b4472138538fd9270e8e60cf6","revisions":{"39554255d7eb77e57970bdd77c479d385e7a058b":{"kind":"REWORK","_number":1,"created":"2019-07-19 18:25:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/671824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/24/671824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/1"}}},"commit":{"parents":[{"commit":"3dea69dc18b59711bb1eaaeec85f64cd6db61b12","subject":"Merge \"build: use upper-constraints.txt from rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/3dea69dc18b59711bb1eaaeec85f64cd6db61b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:25:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:25:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3c4b2fab82eb6047b4ce37119c1af18120234e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/39554255d7eb77e57970bdd77c479d385e7a058b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/39554255d7eb77e57970bdd77c479d385e7a058b"}]},"branch":"refs/heads/master"},"f5ceb57e33a7d52b4472138538fd9270e8e60cf6":{"kind":"REWORK","_number":2,"created":"2019-07-20 09:37:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/671824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/24/671824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/24/671824/2"}}},"commit":{"parents":[{"commit":"3dea69dc18b59711bb1eaaeec85f64cd6db61b12","subject":"Merge \"build: use upper-constraints.txt from rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/3dea69dc18b59711bb1eaaeec85f64cd6db61b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:37:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:37:14.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3c4b2fab82eb6047b4ce37119c1af18120234e43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/f5ceb57e33a7d52b4472138538fd9270e8e60cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/f5ceb57e33a7d52b4472138538fd9270e8e60cf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
