)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~689522","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I6ae29f8a01529bd4786547a11b73171e008dfd0b","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I6ae29f8a01529bd4786547a11b73171e008dfd0b","subject":"PDF Documentation Build tox target","status":"MERGED","created":"2019-10-18 15:28:39.000000000","updated":"2019-10-18 18:28:39.000000000","submitted":"2019-10-18 18:27:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689522-1571423232619-33a7a8e1","meta_rev_id":"5ae1fecf82c8dcadf3a05319340008278ff08e4e","_number":689522,"virtual_id_number":689522,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 18:27:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-10-18 18:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 18:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 18:16:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 18:12:40.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-10-18 18:16:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-18 18:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"190471801f901a79fc6b19c3ce36cce93ff40de9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 15:28:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c2c749622b0e04bbcc0dcb175e6ce747492521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:41:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a44a35712a34de899e78f85ea6381c6 : SUCCESS in 2m 24s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3fa4e6bb5b6a4833bbf8378c69f8d266 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72771e2ab25a4361811e95e9c41f2c05 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98614a8dd3f24320992c448d6f1a14a5 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"97f365b870d04f21b363e8ce3fb6253ccc66c039","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-10-18 18:12:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1945cbf5e8df975856bbf627525e7422baafa97","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 18:16:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"999e2e1e4830964090ec07c878bab51733fdc79f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:16:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8111bdf15907080a9d998f549c7da04b3540659","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:27:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f229d60f722b7f52585ccab58b29ef783decc37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:27:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cd68005f264479c9562c85a1a9f0e8a : SUCCESS in 3m 11s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e11581b5b5041db83837c9cf8df38b9 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d2591fd29c4ae2b031b4295f30738b : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00e40c4e6e4a43b3a9b0384d7a195fa2 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7676384ca2126e942f482d3bd6b6738c179eb377","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:28:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c286f67a28a046ee81f2cefb1fbbf5d3 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40aa143b7914f53acc9fcf23628591e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e31110da3c2bbd520b94c4a9c2d454b00f8b268","revisions":{"2e31110da3c2bbd520b94c4a9c2d454b00f8b268":{"kind":"REWORK","_number":1,"created":"2019-10-18 15:28:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/689522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/22/689522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/22/689522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/22/689522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/22/689522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/22/689522/1"}}},"commit":{"parents":[{"commit":"cae0d4f431a15ce7e95c62518b610bd3ca2da6b2","subject":"Allow venv python interpreter to be overridden","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/cae0d4f431a15ce7e95c62518b610bd3ca2da6b2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 15:28:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 15:28:36.000000000","tz":180},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: I6ae29f8a01529bd4786547a11b73171e008dfd0b\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2e31110da3c2bbd520b94c4a9c2d454b00f8b268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2e31110da3c2bbd520b94c4a9c2d454b00f8b268"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
