)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~694349","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I73887784cbc3ce6089a42c67871dbc6c3d54acf6","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I73887784cbc3ce6089a42c67871dbc6c3d54acf6","subject":"Change upper constraints url","status":"MERGED","created":"2019-11-14 16:34:16.000000000","updated":"2019-11-26 22:41:44.000000000","submitted":"2019-11-26 22:40:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694349-1574808012885-8eeca06d","meta_rev_id":"0db3616c9286d3ca50283a8d9b3b36deaef2204e","_number":694349,"virtual_id_number":694349,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-11-26 16:44:29.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 22:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-11-26 12:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 09:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-11-26 12:40:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 09:29:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-26 12:40:50.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-26 22:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8cf77efbdee6307802048b44a13c6429ddb12fc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 16:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75b1409395d816c454c47af6cc0f07d5b1ae1a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:05:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa3f4e0681b44a20be76e530bc752d4c : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/763b1e4b980f40228e33723040257dce : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/341057f249f44de1a28dd29af9a21251 : SUCCESS in 1h 45m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b67874922b6d4d54a8beaa3618ad6063 : FAILURE in 23m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a81d9ccf8dd430dbfc3f62db97dc656 : SUCCESS in 1h 17m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3a9ceb5855254aabb5a96065f09aad69 : SUCCESS in 2h 24m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd55cb5a2a444242af298c3c4aa63745 : SUCCESS in 1h 45m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f456468b2bd346179a072b12c6977ed4 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3961905317ed7879f3465df1419b08cc5f14424d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-17 12:12:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"221e4d99f974483b456b5a11688a54647178677c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 14:36:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/098563490c104fe88cf28afeecbead9f : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1b606bb07eb04dc7a46f03de759e3b08 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6e417f9ea8d24780adcfd532c039c52d : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bda06c3fefa9429ebfed54d415f70f52 : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb37760f5f024973bb9210fec6c8a52d : SUCCESS in 57m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de38aa0dd259462c8fb5afb97dc48627 : SUCCESS in 2h 22m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd9334fcddf84632bd72781cf6799f93 : SUCCESS in 1h 35m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa58a8e47b4465e8fde0f2ce66fed6b : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c604925518501a84f3290dba7b84f90da7902e1a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 09:29:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21b04e708453181f1afe72699b4eddc6214ff0cc","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-26 12:40:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fdd56cdb6bef4ffcb177b234089aa5c999329163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:41:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e055c48542cd835a440e2850bce2408abbc8afd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:19:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d061661b6e564599be5656580a16f953 : FAILURE in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c99e2a5e71fd4281800c7429428eba12 : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c1b8cd25c04a40fea7f690ff24c041db : SUCCESS in 2h 21m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/06cd21397fea4485a56a985546f948d7 : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9efbb18b6b7e4cbe9e08bd7bab6e9e70 : SUCCESS in 1h 08m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d95de1077a834d8f9a88c674813e2349 : FAILURE in 6m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f21d63bfa0f84011ba22ccdf66a6c9a4 : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc64b222a024ff8987efe85ecaad571 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c6f0c350d2a2f4b965f54a8c5f96c12ecf4a7c2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-26 16:44:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35eb2805c56a1ee5cd40842c82b6b53d3fd6ad93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:36:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4205b69f1be043d696b59794fcc4a3a9 : SUCCESS in 4m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9cf97fa08d0847918489f73306c1f3b4 : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e517e1783044ce799551cb2be71f463 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/49bfd073a3c342dbb7745cac38d70081 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/096aad5c52a94be2b141005c0565e78f : SUCCESS in 1h 01m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a9c3bbe08f04c75b271ae6156971b6e : SUCCESS in 2h 46m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65ae4dd104c94ac49c712df713e05f24 : SUCCESS in 1h 41m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9113b8c2e8425b857e4ba51bf86945 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d72a78186915e6566378b4e0b3efae88ac3cfd57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:37:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6810e91d2151bcc98a6841c2228b2e0698227e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:40:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d41901731aff449593a1771e221a88e9 : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/64350239f0a24eb5b6a88a28e9af7d20 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2700fad97735468dbe85fdf706aa5a10 : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dda6a1ff29cc47a8b7de1632a82419d8 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92e81f0576504856ab6b2737e308e6d2 : SUCCESS in 1h 08m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99588bbdcaee42f1ad2db2ebf725ddce : SUCCESS in 2h 57m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b961f80b8bc24d5aa398fc95d376253c : SUCCESS in 1h 53m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea8fd7a1a704293b8aa69f80b4451e2 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a1054368c28d54e1dbb1cfd36900b4089c36c02d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:40:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"413c0bdfd4557403c3a0d855dbe8a7d57dbb1aea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d873766794542ca8a88a0b59bebef58 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b00e91e9c3b485c08aab2448bf1193bf06289b6c","revisions":{"b00e91e9c3b485c08aab2448bf1193bf06289b6c":{"kind":"REWORK","_number":1,"created":"2019-11-14 16:34:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/694349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/49/694349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/49/694349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/49/694349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/49/694349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/49/694349/1"}}},"commit":{"parents":[{"commit":"16b381d097f2273ebbcfa00ab7294f909f3f9735","subject":"Drop common-db tag from db_sync task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/16b381d097f2273ebbcfa00ab7294f909f3f9735"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 16:34:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 16:34:04.000000000","tz":120},"subject":"Change upper constraints url","message":"Change upper constraints url\n\nTo be consistent with other projects, we are changing upper constraints\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I73887784cbc3ce6089a42c67871dbc6c3d54acf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b00e91e9c3b485c08aab2448bf1193bf06289b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/b00e91e9c3b485c08aab2448bf1193bf06289b6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
