)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~695565","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I588ba1e26f5e4afc40f43547b15b48e9428e2f48","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"osa-train","hashtags":[],"change_id":"I588ba1e26f5e4afc40f43547b15b48e9428e2f48","subject":"Removes reference to deprecated rally-manage","status":"MERGED","created":"2019-11-21 20:05:35.000000000","updated":"2019-12-12 14:12:36.000000000","submitted":"2019-11-26 19:11:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695565-1574795480201-81677534","meta_rev_id":"a0f9b2887dae3c501180730878b21a086b78285a","_number":695565,"virtual_id_number":695565,"owner":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 19:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-11-26 16:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-22 20:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-11-26 16:40:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 20:39:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-26 16:40:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-11-26 19:11:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18138c135ced773dd24eb30a8f93a3331e9929fc","author":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"date":"2019-11-21 20:05:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55151b627a548c2133035fde9c0f418cb6ea5a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 23:13:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/af8227ab737e4bb697ca26a46fb8c5ac : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f66769d780bc425db5e05d358903a95e : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2b0dd9765f3c407d92cda34ef2f7cc33 : SUCCESS in 1h 32m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8a695f71275e4ccb9195259aa23f9868 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cb793da8959425d93f5c1d1c5ee8fe3 : SUCCESS in 1h 04m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4123097de844d55b16ea4b63a881e1d : TIMED_OUT in 3h 02m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed2bc63577fe411f93b2fd9b24b69902 : SUCCESS in 1h 58m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f871d015859b423e9e9ce7e06596eac6 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6cc5504761c16d9b89fb69cc4ec2bd82390509dd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-22 07:56:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"800fc39f665d77d85b98db872a3c2a67a3763df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:33:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f39c3e0596b24b03876318f61c602aa3 : SUCCESS in 5m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/46a4e0d9ae974b93a4236a58c7270b13 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/be768dedac5a4a8d96821d20f93d36be : SUCCESS in 1h 45m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8f3dfd46615c43779ad5ca80f3fe5b85 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/140d87b6bb5d4abc898dbb97a942fc60 : SUCCESS in 1h 13m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1874837d3d2541a487794f30a1eb0b56 : SUCCESS in 2h 30m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/952eaee4c5c14ce38138f4327a037662 : SUCCESS in 1h 55m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2510db4d7f734a9daa942972fb4bd317 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c96797cebb1c46b04c11dad6922f584cbac21ed9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-22 20:39:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"658045ba4edf880065fee2c9afd949b0b67d34c5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-26 16:37:51.000000000","message":"Topic changed from bug/1853509 to osa-train","accounts_in_message":[],"_revision_number":1},{"id":"44336cecaa715a2a045fab5d11e8c28c5f14fed3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-26 16:40:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e457d48e2569446abaab6e4c561132e91650aaf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:40:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5e2b61c56df4261b36f3a9ce7221fb359fd6adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:11:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0fbe9fad3e1e4aefb59d054ed092ff73 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fe7093a2128f4b77900c3c2d8262df34 : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a56e889e255147169dc969f33ce99066 : SUCCESS in 1h 33m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/402c038fa8684b64ba6aa2153f1a3e76 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43c97078e6f44833b0015564c6c5697d : SUCCESS in 1h 13m 42s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/942670013112444290a5d488ed985e43 : SUCCESS in 2h 28m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2444b53749a40778690e1ba3ba4d867 : SUCCESS in 1h 58m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23498141e14d4916b5a4d4ad4075afbd : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2f1a427f8dad2d0d9b63c6afed91696af65b0a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:11:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10b4013a51821046fbcc20f59d1354317d6f755c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:12:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e53a2aae0064d56a04c2f2598d43b46 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7607aeb5f3b5f75eccec019deb512579a90b9e9a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-27 11:17:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1d9448767c9ffa3a3d317bee0463e614a98457f5","accounts_in_message":[],"_revision_number":1},{"id":"a0f9b2887dae3c501180730878b21a086b78285a","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-12-12 14:12:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1af78bb2b6331cb1a79c01660e4c04b7df50e039","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fc259310fceac9c3ff8af149ff95484391b5b2b","revisions":{"7fc259310fceac9c3ff8af149ff95484391b5b2b":{"kind":"REWORK","_number":1,"created":"2019-11-21 20:05:35.000000000","uploader":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"ref":"refs/changes/65/695565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/65/695565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/65/695565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/65/695565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/65/695565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/65/695565/1"}}},"commit":{"parents":[{"commit":"4d5a4122f76965b26e787ab834101748525924bb","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/4d5a4122f76965b26e787ab834101748525924bb"}]}],"author":{"name":"Jeff Albert","email":"jralbert@uvic.ca","date":"2019-11-21 20:02:50.000000000","tz":-480},"committer":{"name":"Jeff Albert","email":"jralbert@uvic.ca","date":"2019-11-21 20:04:15.000000000","tz":-480},"subject":"Removes reference to deprecated rally-manage","message":"Removes reference to deprecated rally-manage\n\nReplaces the call to \"rally-manage db create\" with the\npost-2.0.0-compatible \"rally db create\".\n\nCloses-Bug: #1853509\nChange-Id: I588ba1e26f5e4afc40f43547b15b48e9428e2f48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/7fc259310fceac9c3ff8af149ff95484391b5b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/7fc259310fceac9c3ff8af149ff95484391b5b2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
