)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~727416","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~If8017854d98fb0fc071d6afe23f6111da25aa8e4","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"If8017854d98fb0fc071d6afe23f6111da25aa8e4","subject":"[ussuri][goal] Update contributor documentation","status":"MERGED","created":"2020-05-12 19:59:02.000000000","updated":"2020-05-16 07:48:47.000000000","submitted":"2020-05-16 07:47:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727416-1589615248665-be6e565f","meta_rev_id":"3b7a95f8c209e6a0d8836d1a1ae3c89be776c33a","_number":727416,"virtual_id_number":727416,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-16 07:47:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-05-15 22:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-16 07:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-16 07:35:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:36:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-05-16 07:35:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-16 07:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4388154581b2f4d1eb4604a7b51c874a1a999241","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-12 19:59:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b314c8f3e864ed888da00b0168c00b43b0bf8f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 21:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3c45aa46adf450fb43e816680b6cc81 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db4e20b6f6e4d67a93ff4527ef72bd3 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a985ebfcee95a47ac3eed30555049c5cf99cf27e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-15 22:36:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c108769a38742ec2afd6238a51fef1cf7329eeeb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-16 07:35:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c56818e3d55977b7e77eb1eae041db1442fb980d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:35:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2e8449a0e0f484ff74bb4e7b379cc0efcc20e28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:47:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a4494dbeb517bcfd6c0e56c912f5ca07944c21f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:47:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4314631453d4a90844a500b398da82c : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abbbed2bc489423bbf96e032c971f0e3 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6b878947cad5410ee86652b7e9d9680560a8c712","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:48:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec474da2caf045a8bc64d765a14e81d5 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aed41415f09932d6bcc59e42fcd5da663ece47f2","revisions":{"aed41415f09932d6bcc59e42fcd5da663ece47f2":{"kind":"REWORK","_number":1,"created":"2020-05-12 19:59:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/727416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/16/727416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/16/727416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/16/727416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/16/727416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/16/727416/1"}}},"commit":{"parents":[{"commit":"dbce50bde0605ee4c78952de3852bd1b389f321e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/dbce50bde0605ee4c78952de3852bd1b389f321e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 19:58:49.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 19:58:49.000000000","tz":180},"subject":"[ussuri][goal] Update contributor documentation","message":"[ussuri][goal] Update contributor documentation\n\nThis patch updates/adds the contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\nStory: #2007236\nTask: #38554\n\nChange-Id: If8017854d98fb0fc071d6afe23f6111da25aa8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/aed41415f09932d6bcc59e42fcd5da663ece47f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/aed41415f09932d6bcc59e42fcd5da663ece47f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
