)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~735233","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~Ibdbfab5d3ab2eb301a583db469f7a96b8785bf7d","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"interpreter","hashtags":[],"change_id":"Ibdbfab5d3ab2eb301a583db469f7a96b8785bf7d","subject":"remove ansible_python_interpreter","status":"MERGED","created":"2020-06-11 19:59:16.000000000","updated":"2020-06-17 13:11:47.000000000","submitted":"2020-06-16 11:31:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735233-1592307064162-3c0ef9a6","meta_rev_id":"6bf29b7dfc53c369fac783710224f59c363207f0","_number":735233,"virtual_id_number":735233,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-15 15:29:27.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 11:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-16 09:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 19:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-06-16 09:53:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 19:08:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-16 09:53:14.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-16 11:31:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19fd4eec4564a89d828fa3215c6f376facd57c58","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 19:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9709f42392e062284837382833230cc72cf6079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:56:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c1b783bd0a7941f598e1881a516a95a2 : SUCCESS in 11m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9db7722a97aa42e0b59b079f02c7de1f : SUCCESS in 1h 45m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b22fd8ea52df4de9af2afc560ac9b58d : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3032db4af65441290eaba4731d81346 : SUCCESS in 1h 20m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7aedef5a2474453f8135997c8d9b90c3 : FAILURE in 1h 30m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64976d8fc9be453e99742f1156084a7e : FAILURE in 1h 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d2789c0fb74e8d9d69dedb7e8ba7dc : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"86b00260c91e294bffc07a987b19fff9abdb5ee6","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-15 15:29:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c46e8cc914db560d4b5a94c6f7cb5ea8206eac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:33:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5e1bdddf9eb4924874aa58982c3e6b4 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4f2cb329b2c04eb99050a3d278b9769c : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2ac1f2caf763474bb1ce96045e553a5c : SUCCESS in 1h 03m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57af8c2f8cec40ecbef6f08dc46089ab : SUCCESS in 1h 17m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a03b72135c764947b363b02ab21e9e01 : FAILURE in 1h 47m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45bd49cdc873477eb845f7b1f03b5f64 : SUCCESS in 1h 39m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d8e82865bc469485eb0c36787944f6 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"30a0e6d2f787d2b97a39711b79bd541f41e5a46a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 19:08:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab494ab8c0a4fd871ec629320808bd4035b9f14c","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-16 09:53:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b26c62d3fdb817dfa3f4f31fef6197fec11bb20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:53:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ae787622961ee00f677a6737f825c0daa21f87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:31:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1b4368a162334735822f0fdd227d2076 : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1880d12bca6f49e8a8766e969dfd1eac : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a392fbe2910540608845bc2093138359 : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abbcabc5e7bb4e6094026474c6fe3211 : SUCCESS in 1h 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334f088de4bb4295bc1ba8e72050c2eb : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e545c00deac14a84f9106c79c317d757e9dbd608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:31:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2bdfbd46efe143c8dbfe7a21b8b3520f01725a0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7103f0570fe40d6ba9ebb5ba5caaea0 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6e14acbca2fa05b5101822e7e891d3bccf55b5e3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:11:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 41f0570dc8fc46c532379bb7ae995240644bbc48","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2acd385d305381abbdaebf0ca29a9c729347f8d1","revisions":{"2acd385d305381abbdaebf0ca29a9c729347f8d1":{"kind":"REWORK","_number":1,"created":"2020-06-11 19:59:16.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/33/735233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/33/735233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/33/735233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/33/735233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/33/735233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/33/735233/1"}}},"commit":{"parents":[{"commit":"639ef054b688645869af30b729c4f1df0309fe4a","subject":"Merge \"Use newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/639ef054b688645869af30b729c4f1df0309fe4a"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 18:01:20.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 19:59:16.000000000","tz":-240},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: Ibdbfab5d3ab2eb301a583db469f7a96b8785bf7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2acd385d305381abbdaebf0ca29a9c729347f8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/2acd385d305381abbdaebf0ca29a9c729347f8d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
