)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~848666","triplet_id":"openstack%2Fopenstack-ansible-os_rally~master~I8100dfd3544c0530815fb877b44b960d404f635b","project":"openstack/openstack-ansible-os_rally","branch":"master","topic":"osa/continue_zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-07-06 10:35:34.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-07-06 10:35:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8100dfd3544c0530815fb877b44b960d404f635b","subject":"Control rally-openstack installed version","status":"MERGED","created":"2022-07-04 19:29:57.000000000","updated":"2022-07-06 10:37:36.000000000","submitted":"2022-07-06 10:35:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848666-osa/continue_zed","meta_rev_id":"c3ffe3bfb76fc39229a72299100461857bfc0d21","_number":848666,"virtual_id_number":848666,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-06 10:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-07-06 07:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-07-06 09:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2022-07-06 09:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 21:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 07:33:18.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-07-06 09:06:26.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"55332b722bad090a458333b8bd4b5fb27e73473b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-04 19:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bd3264afca2c15d573a7f0e8327342c5380de19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-04 19:33:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eaf2ac4393eaf2847a4c191b24c085ec8ad9be51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-04 19:34:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c2ffeab2672fabcefecbb4aa13115891fe15c92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 21:06:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e875923bd2db40ba9e84e1d2b5e7c1c6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f87f2cf8d5ec4cf78442f8a10a851b64 : SUCCESS in 12m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8cf9f1727d5e47f7bc53271488311270 : SUCCESS in 59m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc5484a1a35b46459fffe4744e1ac7c6 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/43a35c472a6a4e55a4407d4f32c9e424 : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/146c7ccbcaa24ac597b79c9fd06df103 : SUCCESS in 1h 15m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f17c2fb8b2e4f90b0cf3e3ad0d7f423 : SUCCESS in 1h 05m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d8c097fbd4ef4dd48ea717b7b0720aa5 : SUCCESS in 1h 25m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/85bb371a766f4fb4ac637b3bce201c0d : FAILURE in 1h 01m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75d0221b7aaf4988bd1f9121536195e2 : FAILURE in 55m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb2cfa5eb99454e8ebbbf260910536b : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5103507c6db249839d3d69b66b3f886c : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"5559d6cf8df3471c2840daba70c16dd076afac84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-05 15:14:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"82077b95e877b0b93231eb61a4334e0563382b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:19:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88f23be4208400caa0e0e01d27f4bd5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e24b940308144a4b0261d995c7b7c6a : SUCCESS in 18m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c3981b2deac4f0989e4e8bc1fb10a63 : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d080e544f274437adaee26b92a0286a : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e93e8ecf4a24a57a05b2cd2ec2fe232 : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d5c12b8b4447e382e952b1d4c6314e : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11442995ae3b4b5180654d60c4e7be1e : SUCCESS in 1h 19m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6ea98327fef7489f827522b6c5c4f691 : SUCCESS in 1h 36m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/17a30793b5164d5fa55b7506e0e39404 : SUCCESS in 1h 59m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f50e814fc24e476091eb8bbd2f3fc17f : SUCCESS in 2h 00m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012519a202d34870ad1a18bff80fb60d : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55ac06714c0c497d815412680c387bff : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"0094a6c670fcc9c9cb9821871fe6b352f6c61fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-05 19:27:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3e4d90a51121b86b7b5117087f7ff96eb1aa72a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 21:07:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7be8f63a7646acba2b6e04aee461a4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59464ef9ef664dbf9d97ab99a6bf0f84 : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c7fcfae3d8ce4ca983ed6a140c047e6e : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4585fffdc6d41d09e3eba5b45c02ebb : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2d5088a2ee84988a1118993052b7dca : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3efa91640b4618940924992f256bda : SUCCESS in 1h 27m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/530750ef28364c7690c0e27f9017de7d : SUCCESS in 1h 21m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/634517a07f8643db999fed2bd4a32d3d : SUCCESS in 1h 31m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5362b2485d745aba78c5e1b7d335572 : FAILURE in 1h 14m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36eaf9b9c764454fa69ec1708dbf7399 : FAILURE in 1h 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e4c744be62458d9347e43b1aa709fa : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3dc7adf46245c5919f537c6db114aa : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"08c884702a8d193230fc123abefe9e953859a54f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-05 22:03:20.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4e919fa105d41a02b011f1df36078ad41eb3c77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 23:25:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e462f170cc4a45bbd35c8b8ef6ae33\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f48b7a03369a479e94c6da5662877fcf : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3e516c494a17453cbcfad1e08643d4a2 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7c9ff56257240188c4c98398c9606ea : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2ebc3e1aad349beb2699f57c4a6198a : SUCCESS in 1h 07m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26a793acc0834a46a6d9cf106b19dbe1 : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0ccfc06e7034ea88fcc2ececf49791a : SUCCESS in 1h 05m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4931ad80b90e410d8914020c2a95ada9 : SUCCESS in 1h 13m 26s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/74ba52544e0544d38a4e1fda4d305c13 : FAILURE in 1h 02m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa18d6a3e09045d7a88d710e830d0b30 : FAILURE in 52m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6dc5ec3b2d4b29b2c399cbff080aa7 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30974793b85b4848b88f886cf51b1cfa : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":6},{"id":"16683e384fe4660adf08937b41c1afa03ca5164c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-06 06:44:22.000000000","message":"Topic set to osa/continue_zed","accounts_in_message":[],"_revision_number":6},{"id":"ac8c3a2359a74069105e1e363c84294910403bcd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-06 07:33:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5e1ad94fe4437720d589c5c5004f0529ba0d5dc2","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-07-06 09:06:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e3668d6246fbd103fc52fbf586b2ed4525cf5d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 09:08:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"94a8995346e37b0b15809d02336f521573945287","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 10:35:34.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0a32e643c394780ae730edace125a34\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4f4a50ccc7c4e81afee28da7ee1ef79 : SUCCESS in 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c02fb15da7f346d5b68fb4cfb9b67219 : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a324722afe9d443489b0f83625183ba8 : SUCCESS in 1h 20m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e21cfac7b00458191d388b5abc7df1e : SUCCESS in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ec1dd5fced148719fe70b5c4ed6d13b : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8b8aab815cfe4bd1ae5bebd043824b35 : SUCCESS in 1h 22m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08519445ba84e4e8d4000e605e2bee7 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb0497dcdb14b4c90b2759b386fdb90 : SUCCESS in 2m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7facf51da04b422bbd1263db008ad98e : FAILURE in 55m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53985ff70615b3ad12558aae43130170e64d7d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 10:35:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c3ffe3bfb76fc39229a72299100461857bfc0d21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 10:37:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38c527fae8724ab4a90ffac2f388d4ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f257d33c4fa41ce82ec5ba23cbf4bdb : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"136dfa02d050454d3b663186a212184a86b63c3e","revisions":{"067d451020dc7700e04ba4f12eb9de452cb7dc2e":{"kind":"REWORK","_number":1,"created":"2022-07-04 19:29:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/848666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/1"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"subject":"Controll rally-openstack istalled version","message":"Controll rally-openstack istalled version\n\nAt the moment we don\u0027t have control over what version of rally-openstack\nis being installed. So we were installing latest version available in\nPyPi. However rally package is constrained in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-penstack as well as installed version.\n\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/067d451020dc7700e04ba4f12eb9de452cb7dc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/067d451020dc7700e04ba4f12eb9de452cb7dc2e"}]},"branch":"refs/heads/master"},"890e1dc972fb473f62d07d85c32b7100b4c893fe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-04 19:33:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/848666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/2"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-04 19:33:45.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/890e1dc972fb473f62d07d85c32b7100b4c893fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/890e1dc972fb473f62d07d85c32b7100b4c893fe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"06140855aec95e7543da89cce898cc6755cb1269":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-04 19:34:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/848666/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/3"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-04 19:34:00.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/06140855aec95e7543da89cce898cc6755cb1269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/06140855aec95e7543da89cce898cc6755cb1269"}]},"branch":"refs/heads/master","description":"Edit commit message"},"48862260f5aa96fa98075413f17322d25d6fbe5e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-07-05 15:14:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/848666/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/4"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-05 15:14:09.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/848760\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/48862260f5aa96fa98075413f17322d25d6fbe5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/48862260f5aa96fa98075413f17322d25d6fbe5e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d7d5b0cf8ce4ab5d27816d1f2ee9cf79643df1d3":{"kind":"REWORK","_number":5,"created":"2022-07-05 19:27:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/848666/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/5"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-05 19:27:38.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nWe also temporary disable upgrade jobs to resolve circular dependency\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/d7d5b0cf8ce4ab5d27816d1f2ee9cf79643df1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/d7d5b0cf8ce4ab5d27816d1f2ee9cf79643df1d3"}]},"branch":"refs/heads/master"},"136dfa02d050454d3b663186a212184a86b63c3e":{"kind":"REWORK","_number":6,"created":"2022-07-05 22:03:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/66/848666/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/66/848666/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/66/848666/6"}}},"commit":{"parents":[{"commit":"854b8a2b1f35e441cc09fa6b51b8b586351a4a53","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/854b8a2b1f35e441cc09fa6b51b8b586351a4a53"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-07-05 22:03:20.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nWe also temporary disable upgrade jobs to resolve circular dependency\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/136dfa02d050454d3b663186a212184a86b63c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/136dfa02d050454d3b663186a212184a86b63c3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
