)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~848718","triplet_id":"openstack%2Fopenstack-ansible-os_rally~stable%2Fvictoria~I8100dfd3544c0530815fb877b44b960d404f635b","project":"openstack/openstack-ansible-os_rally","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-07-12 08:34:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I8100dfd3544c0530815fb877b44b960d404f635b","subject":"Control rally-openstack installed version","status":"MERGED","created":"2022-07-05 15:08:48.000000000","updated":"2022-07-12 10:00:32.000000000","submitted":"2022-07-12 09:58:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848718","cherry_pick_of_change":848666,"cherry_pick_of_patch_set":3,"meta_rev_id":"edd5902d40f18e7a9e026f53cbf37a5ce8267ec4","_number":848718,"virtual_id_number":848718,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 09:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-12 08:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-12 08:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-07-12 08:34:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-05 16:54:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 08:21:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9e7c11269b0a0008c6d2e99c79b561cad071a2fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-05 15:08:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ada7caf6b07d03e261f87a3fdcbc359aeba8d91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:54:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57dc5e6ceb1a4704a054cae98d6fd76f\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/508a3279e4a749da97d654e225462778 : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/387d7d03ae8d4cbd8517ba8f7af52299 : SUCCESS in 1h 13m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7579228a1a214bb78d98ee2e22dcdf48 : FAILURE in 29m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4369d73bbf9a41c984dcea95553e6cb9 : SUCCESS in 1h 10m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa6c17255ee94ea6b51728ec8592a12d : SUCCESS in 1h 41m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3033683564df4defb9ab00955b957be7 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/625fe2e16a4e4aa38af0ec04ffa6a391 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"67e5ffa8073ca0e8eb28a317b2be4de3e6826a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-12 06:17:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"302f4af9a8e6e16cef4378d233c1f361d91c465b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 08:03:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c2a45618df4fbb9594fe7641ea7765\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6bbf71b1cd6a467d831313d8616a30a9 : SUCCESS in 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7a0da3c7edeb49bca4016301b63b2d6f : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2ea2185f6c04669a6e12bcaf7463bd9 : FAILURE in 34m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee58759a05f245ff81014e37c0a8843c : SUCCESS in 1h 15m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f9545a50b124c5c9b0430e220388249 : SUCCESS in 1h 39m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c850976deb6144459b93bed955de3618 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab71895bd0c34c0d8d5e9e1e3d3f6c46 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"bbdc24740c66898efa2ba4b297d5f60af08dae3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-12 08:21:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5b44f0ab7e6f0eba7f3a2208c1d8149f5d2f5c0f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-12 08:34:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"833164b52cc12260bd58346d1889d360ec3afd68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 08:37:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c35cf9e87ba27aca318c8fb361f31a03c4820029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:58:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afeb18775a542579563140bdc9afac4\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2b4052f275194b7ba001cd9c9c10ffd8 : SUCCESS in 10m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/04de5008767e4d7ca16ec84816b595c2 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1008257ac0744632b1af2a323d157e93 : FAILURE in 28m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce7962c32df4fbea22080143d92f343 : SUCCESS in 1h 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de4c0ffc6834a85a80169d30929ce60 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a330a5966654060a0d45f546b68db8e : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8b8377ddbbfa071bdd2b12238256a86e4f2ee11d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:58:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"edd5902d40f18e7a9e026f53cbf37a5ce8267ec4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:00:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0357c7c1ef454f8e6c5f7619cd2767\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4acf161cf2c44900a22f741e850bc7f9 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/523087051420456cb743526050202f54 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"401ab742d09ddc64005c7af8809d337568338e27","revisions":{"ff4e967361e83f6a73b6ebbe47d80759a4548163":{"kind":"REWORK","_number":1,"created":"2022-07-05 15:08:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/848718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/18/848718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/1"}}},"commit":{"parents":[{"commit":"138b950afe681b637757fe6f5001d6b8d06539cc","subject":"Ensure that the rally venv build is self contained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/138b950afe681b637757fe6f5001d6b8d06539cc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-05 15:08:48.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/ff4e967361e83f6a73b6ebbe47d80759a4548163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/ff4e967361e83f6a73b6ebbe47d80759a4548163"}]},"branch":"refs/heads/stable/victoria"},"401ab742d09ddc64005c7af8809d337568338e27":{"kind":"REWORK","_number":2,"created":"2022-07-12 06:17:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/848718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/18/848718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/18/848718/2"}}},"commit":{"parents":[{"commit":"138b950afe681b637757fe6f5001d6b8d06539cc","subject":"Ensure that the rally venv build is self contained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/138b950afe681b637757fe6f5001d6b8d06539cc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-12 06:17:18.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/401ab742d09ddc64005c7af8809d337568338e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/401ab742d09ddc64005c7af8809d337568338e27"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
