)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~849229","triplet_id":"openstack%2Fopenstack-ansible-os_rally~stable%2Fxena~I8100dfd3544c0530815fb877b44b960d404f635b","project":"openstack/openstack-ansible-os_rally","branch":"stable/xena","topic":"osa/continue_zed-stable/xena","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-07-12 22:25:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8100dfd3544c0530815fb877b44b960d404f635b","subject":"Control rally-openstack installed version","status":"MERGED","created":"2022-07-12 10:45:46.000000000","updated":"2022-07-12 22:26:50.000000000","submitted":"2022-07-12 22:25:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849229-osa/continue_zed-stable/xena","cherry_pick_of_change":848666,"cherry_pick_of_patch_set":6,"meta_rev_id":"6a73abe6b00028f31bbe08a1141358b6e0521819","_number":849229,"virtual_id_number":849229,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 22:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-12 20:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-12 21:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 21:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 16:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 21:03:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0e8f779b83f50a0b4fdb5a2ba728281bf09cbb8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-12 10:45:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dc87797463738546579216a4a56675f74a7a408b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:13:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5698660f9b0c43379a38a9632706a815\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cadb14d0c474646bd149c58b87a0796 : SUCCESS in 16m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/724eb5cee815465f9a3ba50fa86e9694 : SUCCESS in 1h 43m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8095738280464b408dec5e490a7dd12a : POST_FAILURE in 1h 03m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2330c28d3d344061ade088e094df9321 : POST_FAILURE in 1h 37m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41702560e4434e209ec0e94fcfd8cd17 : SUCCESS in 1h 40m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cc95d38d2c248c3bbbbb6dd4394851e : SUCCESS in 1h 45m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6748c05c50c945818c2b02108239caea : POST_FAILURE in 1h 12m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/339a3387803c4ef8ad25d83b27615784 : FAILURE in 1h 03m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12da9387ad5844938b827a9c8677f491 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64fb504fb6934100bb38baf870efc422 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f009582de1b7c449d855585af25579f62031c4c6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-12 16:27:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f68a5569e17f7f45e347c53ff20af71bf887fd7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5139befe4dcd42a7838290f2058ca3e8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f025cd50448843f9893f54462ca0483d : SUCCESS in 14m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1d4782e7842540eaba91871d3397b5e5 : SUCCESS in 1h 34m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/71401cac7344439c9526e71255cf9418 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/341f3ef06edc4fadbc0fb72e428b3dfe : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ec60505e22c4a2896e58df7e2c837db : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/566e347711974b5091952d4125e7cdee : SUCCESS in 1h 12m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c76643f75a7c4b20b6cafbc17c97e627 : FAILURE in 1h 23m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d246ec4e695b400bb83eb26df24a0082 : FAILURE in 1h 05m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7f134d6883420ab5a8304d42a192c4 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eb421a2651d48588d60f1cc5eff70f7 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"738429b1f363206b1de97ef60874ec5201590fb5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-12 20:25:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"718deb60361f676b2959ffdc981622694132c528","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-12 21:03:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5418b94705c7b3daeae7b1ea0349ef09b742099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:06:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"148d28151e81b551a9d2190286defe09d3fe2bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 22:25:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523b6a432f6a4f0ba8a39a336bd188a6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/271fd6c3084f4cae9b8e5b068f18fa5e : SUCCESS in 16m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e722549f086d4efda19aa61c52673677 : SUCCESS in 1h 15m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e42c55cff7e49158b14bddacab8e1dd : SUCCESS in 56m 56s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a590313db0e741f4a78bf9ea52d8c4dd : SUCCESS in 1h 12m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6286bd276aae417fb61024bfa5bc294a : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9309d8c6276a47bf9959a7d3c71d127a : SUCCESS in 1h 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a96910373c43809cf4a5a8131955da : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7c7d197f4664313847815ac17a27f45 : SUCCESS in 3m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bbac49df3764be1a72793d5fab06854 : FAILURE in 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d6fe5a994cf229b618222cfd7472805f0db49c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 22:25:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a73abe6b00028f31bbe08a1141358b6e0521819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 22:26:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5283980f1b0f46f59c6a711f624c05ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27bc2f502ba4cde9d713ae37f302931 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40e47aaab3ccbbab22b49ca7c1a46f15243b7b28","revisions":{"40e47aaab3ccbbab22b49ca7c1a46f15243b7b28":{"kind":"REWORK","_number":1,"created":"2022-07-12 10:45:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/849229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/29/849229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/29/849229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/29/849229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/29/849229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/29/849229/1"}}},"commit":{"parents":[{"commit":"5cd6439985624ac1ef5faf4cbea25664c8224872","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/5cd6439985624ac1ef5faf4cbea25664c8224872"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-04 19:29:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-12 10:45:46.000000000","tz":0},"subject":"Control rally-openstack installed version","message":"Control rally-openstack installed version\n\nAt the moment we don\u0027t have control what version of rally-openstack\nis being installed. So we are installing latest version available in\nPyPi. However rally package version is defined in\nrally-openstack constraints.\n\nWith this patch we add variables to be able to adjust source of\nrally-openstack as well as pick version for installation.\n\nWe also temporary disable upgrade jobs to resolve circular dependency\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/847272\nChange-Id: I8100dfd3544c0530815fb877b44b960d404f635b\n(cherry picked from commit 136dfa02d050454d3b663186a212184a86b63c3e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/40e47aaab3ccbbab22b49ca7c1a46f15243b7b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/40e47aaab3ccbbab22b49ca7c1a46f15243b7b28"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
