)]}'
{"id":"openstack%2Fopenstack-ansible-os_rally~861730","triplet_id":"openstack%2Fopenstack-ansible-os_rally~stable%2Fussuri~Ia9984e072e90bcfca1758d114297b07ad68e082e","project":"openstack/openstack-ansible-os_rally","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-19 17:15:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9984e072e90bcfca1758d114297b07ad68e082e","subject":"Move rally details to constraints","status":"MERGED","created":"2022-10-18 10:20:03.000000000","updated":"2022-10-19 17:17:14.000000000","submitted":"2022-10-19 17:15:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861730","meta_rev_id":"1087c1394b0d6c648c0af122c9c6b1cbb9e8fa3a","_number":861730,"virtual_id_number":861730,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 17:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2022-10-19 13:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-19 16:05:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 16:05:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 12:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 13:34:47.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2022-10-19 16:05:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c507c69af1df6df0f5055418075433cfe4db541e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 10:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a4645a8f65005a3bc0d86604ea2ffc2384277a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 11:52:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"21bc1421c6b56a0bfc51b5b03984477f0b7cf9db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:10:36.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/861742 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c54b65e84b11ae7d35500bae55e3a4a3f4cf2c92","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 12:32:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e75661670a6b86bc7f5c98ef35e5f2629f96adae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:59:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2844144ff37c43fe8e1af48ded26613e\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf8f8ffaf8034f4698fc895c2e6bec3a : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c0e942e9ba3247d0b0ed5f67fee4162d : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca92c7cf91a4428997796d5661ca45ae : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5bf7ed8784f4f2ab930a814fcca42a7 : SUCCESS in 1h 06m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91cad0640607441891f80ca07c08e998 : SUCCESS in 1h 14m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8ace00762b24a3096b2fae1ecade594 : SUCCESS in 2h 19m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57b5711ebc92487bbcbb189d37e57785 : FAILURE in 20m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8b3b2fa6d94ef4a87f783a78a996fc : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e7a02307e3fac443267eb42a13bf9cf5f7536a56","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-10-19 13:34:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3ed4c92b7738db882ab619b04ad519f15e60332","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-19 16:05:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5d10bded5cf5777a3dde649433652ca8f22ab69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:07:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"541d58e71fac46368127a1e57b661e47247a60d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:15:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6100e673c119404a892c317676d54a0d\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/50f4c2d475c24e339ce1472bb9160bb1 : SUCCESS in 7m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f52191ccd094b0ab83bb330f51f40fd : SUCCESS in 1h 02m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3e370dcbad584c16be348d927661dfd6 : SUCCESS in 1h 07m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d5bc4e045814edbac826645c65b4de5 : SUCCESS in 1h 07m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d70347940d4bffb4ebe982ee1be621 : SUCCESS in 1h 00m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f4297588904c5a8ede8676d7e72cb5 : SUCCESS in 5m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fee651d711b467c937e4270b66b758b : FAILURE in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13d5f9201164d83d4a4b10cdf95bbf79ffc44285","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:15:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1087c1394b0d6c648c0af122c9c6b1cbb9e8fa3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:17:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a74131cec41431284995a0eb0277251\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b43ed4001274ecdafe622895a041a3d : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"656ce0b6174d1b9366d6f6a4d8be45d2d585274f","revisions":{"50515db0b7c398495fbccea2cf9a0cbbc8ce059f":{"kind":"REWORK","_number":1,"created":"2022-10-18 10:20:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/861730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/30/861730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/1"}}},"commit":{"parents":[{"commit":"690fd3bdb25ab0a6d6e271bd37c99d386ac94aa8","subject":"Control rally-openstack installed version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/690fd3bdb25ab0a6d6e271bd37c99d386ac94aa8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 10:19:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 10:19:50.000000000","tz":120},"subject":"Move rally details to constraints","message":"Move rally details to constraints\n\nUssuri does have older resolver, thus it can not handle details in\nrequirements, and they should be provided in constraints.\n\nChange-Id: Ia9984e072e90bcfca1758d114297b07ad68e082e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/50515db0b7c398495fbccea2cf9a0cbbc8ce059f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/50515db0b7c398495fbccea2cf9a0cbbc8ce059f"}]},"branch":"refs/heads/stable/ussuri"},"656ce0b6174d1b9366d6f6a4d8be45d2d585274f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-18 11:52:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/861730/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_rally","ref":"refs/changes/30/861730/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_rally refs/changes/30/861730/2"}}},"commit":{"parents":[{"commit":"690fd3bdb25ab0a6d6e271bd37c99d386ac94aa8","subject":"Control rally-openstack installed version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/690fd3bdb25ab0a6d6e271bd37c99d386ac94aa8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 10:19:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-18 11:52:23.000000000","tz":0},"subject":"Move rally details to constraints","message":"Move rally details to constraints\n\nUssuri does have older resolver, thus it can not handle details in\nrequirements, and they should be provided in constraints.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-tests/+/861742\nChange-Id: Ia9984e072e90bcfca1758d114297b07ad68e082e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/656ce0b6174d1b9366d6f6a4d8be45d2d585274f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_rally/commit/656ce0b6174d1b9366d6f6a4d8be45d2d585274f"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
