)]}'
{"id":"openstack%2Fopenstack-ansible-os_sahara~658613","triplet_id":"openstack%2Fopenstack-ansible-os_sahara~stable%2Fstein~I640eb17262633f9219026f7bcd2c7361eef9686c","project":"openstack/openstack-ansible-os_sahara","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I640eb17262633f9219026f7bcd2c7361eef9686c","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:59:36.000000000","updated":"2019-05-12 03:18:28.000000000","submitted":"2019-05-12 03:18:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658613-1557631108552-f41f2753","meta_rev_id":"a90569e86432270ca351448ab31430a17946a027","_number":658613,"virtual_id_number":658613,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:56:16.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:56:16.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:56:16.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45628413b200497ba437d777465dceb6e2aaf119","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4acfefce18bcff30c9c9add3e12a67acea9a6514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/658613/1/check/openstack-ansible-linters/1527889/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/13/658613/1/check/openstack-tox-docs/a24ab5e/html/ : SUCCESS in 3m 21s\n- build-openstack-releasenotes http://logs.openstack.org/13/658613/1/check/build-openstack-releasenotes/3fd103a/html/ : SUCCESS in 2m 21s\n- openstack-ansible-deploy-aio_metal_sahara-centos-7 http://logs.openstack.org/13/658613/1/check/openstack-ansible-deploy-aio_metal_sahara-centos-7/c340d04/ : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal_sahara-debian-stable http://logs.openstack.org/13/658613/1/check/openstack-ansible-deploy-aio_metal_sahara-debian-stable/41bbe04/ : SUCCESS in 1h 00m 02s\n- openstack-ansible-deploy-aio_metal_sahara-ubuntu-bionic http://logs.openstack.org/13/658613/1/check/openstack-ansible-deploy-aio_metal_sahara-ubuntu-bionic/d35e29b/ : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d19baf487745e0918a0bccc7ade3044078b55f0a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:56:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d97079f6534c32ce09442b61ab3ae770ba2d342d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:56:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"927de19241ba93bc4ba0a188fada5186a9f9c4f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:18:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4a74386bccf3a691a6077179a284edb9f0386ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:18:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/658613/1/gate/openstack-ansible-linters/15dced3/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/13/658613/1/gate/openstack-tox-docs/359fdbd/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/13/658613/1/gate/build-openstack-releasenotes/42f4155/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_sahara-centos-7 http://logs.openstack.org/13/658613/1/gate/openstack-ansible-deploy-aio_metal_sahara-centos-7/49c35ba/ : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal_sahara-debian-stable http://logs.openstack.org/13/658613/1/gate/openstack-ansible-deploy-aio_metal_sahara-debian-stable/2bb51be/ : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-aio_metal_sahara-ubuntu-bionic http://logs.openstack.org/13/658613/1/gate/openstack-ansible-deploy-aio_metal_sahara-ubuntu-bionic/645a830/ : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3c3647e8f95b3007209d0c8e58fbb6408495963","revisions":{"f3c3647e8f95b3007209d0c8e58fbb6408495963":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:59:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/658613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_sahara","ref":"refs/changes/13/658613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_sahara refs/changes/13/658613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_sahara refs/changes/13/658613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_sahara refs/changes/13/658613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_sahara refs/changes/13/658613/1"}}},"commit":{"parents":[{"commit":"bd108b1d97c0a4e9df4ef6dc78fff3293584ad55","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_sahara/commit/bd108b1d97c0a4e9df4ef6dc78fff3293584ad55"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:35.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I640eb17262633f9219026f7bcd2c7361eef9686c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_sahara/commit/f3c3647e8f95b3007209d0c8e58fbb6408495963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_sahara/commit/f3c3647e8f95b3007209d0c8e58fbb6408495963"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
