)]}'
{"id":"openstack%2Fopenstack-ansible-os_searchlight~654377","triplet_id":"openstack%2Fopenstack-ansible-os_searchlight~master~I8464eafee3c62a6195282ca9bd124258f2dc5500","project":"openstack/openstack-ansible-os_searchlight","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8464eafee3c62a6195282ca9bd124258f2dc5500","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-04-22 07:16:08.000000000","updated":"2019-04-22 10:56:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e81e286d236f3d2dce4aa17d51341cf6f4b6b29","_number":654377,"virtual_id_number":654377,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-04-22 08:22:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 08:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd1e9e1b6869f8dea2ff4d0e29e0b0854b80c857","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80cde1dec11806fc6b95cfeb630ef916b1ac6848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:22:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/77/654377/1/check/openstack-ansible-linters/0ad681c/ : SUCCESS in 4m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/77/654377/1/check/openstack-ansible-functional-debian-stable/920487a/ : FAILURE in 24m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/77/654377/1/check/openstack-ansible-functional-centos-7/27a4891/ : FAILURE in 25m 15s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/77/654377/1/check/openstack-ansible-functional-ubuntu-bionic/a3dc05a/ : FAILURE in 23m 51s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/77/654377/1/check/openstack-tox-docs/c4d24c3/html/ : SUCCESS in 3m 17s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/77/654377/1/check/openstack-ansible-functional-ubuntu-xenial/b62534d/ : FAILURE in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bff6e4c7ca0758e4f762753daedfba47700181f7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:56:13.000000000","message":"Abandoned\n\nRole is being retired","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cb001aa115426e71a9d763228db593ecbd1d5b1","revisions":{"4cb001aa115426e71a9d763228db593ecbd1d5b1":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:16:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/654377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_searchlight","ref":"refs/changes/77/654377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/77/654377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/77/654377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/77/654377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/77/654377/1"}}},"commit":{"parents":[{"commit":"8750b0aaeda354e9cf33e9d235fb69547981386d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/8750b0aaeda354e9cf33e9d235fb69547981386d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:06.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8464eafee3c62a6195282ca9bd124258f2dc5500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/4cb001aa115426e71a9d763228db593ecbd1d5b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/4cb001aa115426e71a9d763228db593ecbd1d5b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
