)]}'
{"id":"openstack%2Fopenstack-ansible-os_searchlight~654532","triplet_id":"openstack%2Fopenstack-ansible-os_searchlight~stable%2Fstein~Ibc7cf610dea99bdf0d17a987bee13da2f5bedb1b","project":"openstack/openstack-ansible-os_searchlight","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ibc7cf610dea99bdf0d17a987bee13da2f5bedb1b","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-04-22 15:39:53.000000000","updated":"2019-08-28 07:53:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bd80b86258b246ebd92021af3e0752ebfb030b6","_number":654532,"virtual_id_number":654532,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":-1,"date":"2019-05-12 00:34:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:16:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-24 22:17:54.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-05-12 00:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a2229a5ebb99ab25f2f92b9395f1cd06d5ca023","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5acb4f6d47629649cd679ad132c55c97406893f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:16:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0392f42ba4f3072a026d8f129bfa04ef5c2a945f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:47:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/312a2c5/ : SUCCESS in 8m 03s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/0e0fd5a/ : FAILURE in 25m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/c55f1a9/ : FAILURE in 32m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/7807905/ : FAILURE in 21m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/5f567a4/html/ : SUCCESS in 2m 30s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/2e92a60/ : FAILURE in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c93a13d6e9592b44ae33b6d942c9e69a2e746ae1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 17:36:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae4a1e50a806bb7ea077e3e87aee910909132444","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:24:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/3ba6991/ : SUCCESS in 6m 05s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/b2659c3/ : FAILURE in 22m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/d3fdb7f/ : FAILURE in 29m 13s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/33ff32f/ : FAILURE in 22m 31s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/7ea920d/html/ : SUCCESS in 3m 08s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/2ec21b5/ : FAILURE in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc2feeb984dd65a5562943f5e7175f76385876f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 19:13:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9624ecd78a686e1b6200c990b54befe028705b82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:22:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/7268f4b/ : SUCCESS in 5m 51s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/4c5e915/ : FAILURE in 26m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/5997ddd/ : FAILURE in 27m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic finger://ze02.openstack.org/2b21e198364e4c7f8865a0fae08eb40f : RETRY_LIMIT in 14m 14s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/9add24a/html/ : SUCCESS in 2m 11s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/3e43a28/ : FAILURE in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc8689763ba813d149aa4e9850fc5931e3b36da9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 21:18:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2681db1b9fe80888cc927cde7c8ec2fb9c7c9fea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:56:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/2f93821/ : SUCCESS in 5m 27s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/43e52cc/ : FAILURE in 23m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/df0d8dd/ : FAILURE in 27m 31s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/59e80f5/ : FAILURE in 23m 45s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/efe7d8d/html/ : SUCCESS in 2m 34s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/89a7f28/ : FAILURE in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73f285f7a4c27a2382c636c895a8fde151e92633","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 07:12:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95590e74b6576ebcd2374422e292d6ac57119b4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 07:48:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/ccf1ff0/ : SUCCESS in 6m 46s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/96cee98/ : FAILURE in 25m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/9e8b220/ : FAILURE in 28m 46s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/27b184c/ : FAILURE in 23m 11s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/f09c388/html/ : SUCCESS in 3m 09s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/50d808b/ : FAILURE in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3db433215d7612158bd81f67ab56cf90de160aa1","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-04-24 22:17:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72d632719a7f816dc0ca90c956fb2b2a9800ab58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 00:01:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/a969c6e/ : SUCCESS in 6m 15s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/8a0fe84/ : FAILURE in 23m 46s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/a0d3898/ : FAILURE in 25m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/8209c35/ : FAILURE in 22m 32s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/0a73109/html/ : SUCCESS in 3m 35s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/df9b210/ : FAILURE in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feca8ab61f3f21dffdb49ad63947ee6ba03cdbe5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 05:00:55.000000000","message":"Patch Set 1:\n\nIt\u0027s not going to pass, as we\u0027ve added Debian as voting one...","accounts_in_message":[],"_revision_number":1},{"id":"d9c9c01061e01109da32f7ea627fbea075228fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 06:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/1/check/openstack-ansible-linters/1c8041f/ : SUCCESS in 5m 49s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-debian-stable/fc338be/ : FAILURE in 26m 47s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-centos-7/3cdf62d/ : FAILURE in 24m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-bionic/564e510/ : FAILURE in 21m 53s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/1/check/openstack-tox-docs/7a01d66/html/ : SUCCESS in 2m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/1/check/openstack-ansible-functional-ubuntu-xenial/e738fd7/ : FAILURE in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdd942e564e6ddf1185aecaee6eea95e59f7e7f5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99118c7472c242b5056e256aaab597c9314d9f8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 00:34:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/32/654532/2/check/openstack-ansible-linters/a29b8e0/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/32/654532/2/check/openstack-ansible-functional-debian-stable/b475fa8/ : FAILURE in 24m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/32/654532/2/check/openstack-ansible-functional-centos-7/de810b8/ : FAILURE in 27m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/32/654532/2/check/openstack-ansible-functional-ubuntu-bionic/6dbde25/ : FAILURE in 22m 38s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/32/654532/2/check/openstack-tox-docs/bd88324/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes http://logs.openstack.org/32/654532/2/check/build-openstack-releasenotes/6a4d520/ : POST_FAILURE in 4m 07s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/32/654532/2/check/openstack-ansible-functional-ubuntu-xenial/f90c45f/ : FAILURE in 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d18f8497d14e3cfce512137d31b938c6e88dd6f3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-28 07:53:45.000000000","message":"Abandoned\n\nos_searchlight repo is retired","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eef703e170526ad9f8ee1fda362eac05cc9cbf8a","revisions":{"77edf774195f6198f514016b9771d5a4c30ad626":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:39:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/654532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_searchlight","ref":"refs/changes/32/654532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/1"}}},"commit":{"parents":[{"commit":"49741e645aa0fd09cdc523a89faef83991692544","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/49741e645aa0fd09cdc523a89faef83991692544"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:52.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ibc7cf610dea99bdf0d17a987bee13da2f5bedb1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/77edf774195f6198f514016b9771d5a4c30ad626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/77edf774195f6198f514016b9771d5a4c30ad626"}]},"branch":"refs/heads/stable/stein"},"eef703e170526ad9f8ee1fda362eac05cc9cbf8a":{"kind":"REWORK","_number":2,"created":"2019-05-11 23:59:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/654532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_searchlight","ref":"refs/changes/32/654532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/32/654532/2"}}},"commit":{"parents":[{"commit":"49741e645aa0fd09cdc523a89faef83991692544","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/49741e645aa0fd09cdc523a89faef83991692544"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:38.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ibc7cf610dea99bdf0d17a987bee13da2f5bedb1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/eef703e170526ad9f8ee1fda362eac05cc9cbf8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/eef703e170526ad9f8ee1fda362eac05cc9cbf8a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
