)]}'
{"id":"openstack%2Fopenstack-ansible-os_searchlight~664103","triplet_id":"openstack%2Fopenstack-ansible-os_searchlight~stable%2Fqueens~I50092ea057e74a694aaa29874abccc1423033209","project":"openstack/openstack-ansible-os_searchlight","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I50092ea057e74a694aaa29874abccc1423033209","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-06-07 23:28:24.000000000","updated":"2019-08-27 15:32:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"778ea8226120a91d62f58d1f929f8975e29a1e55","_number":664103,"virtual_id_number":664103,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-08 00:03:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-11 20:55:23.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 00:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 20:55:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"04ce37dc7c57b1b34c7189a73a84e7bc0171c99e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:28:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9724cfc3f1a92762d8fdbb8c53abba42d8dccec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:03:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/664103/1/check/build-openstack-sphinx-docs/f385fa9/html/ : SUCCESS in 3m 43s\n- openstack-ansible-linters http://logs.openstack.org/03/664103/1/check/openstack-ansible-linters/7ae7535/ : FAILURE in 5m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/03/664103/1/check/openstack-ansible-functional-ubuntu-xenial/94e3052/ : SUCCESS in 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"24959ac736750472337509e77bb98eaf822396e2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:55:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e89194fa5c842ad78b2e0b34a56025782278b23f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:34:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- noop noop : SUCCESS in 0s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/664103/1/check/build-openstack-sphinx-docs/2de7c4a/html/ : SUCCESS in 3m 47s\n- openstack-ansible-linters http://logs.openstack.org/03/664103/1/check/openstack-ansible-linters/bd2c768/ : FAILURE in 7m 22s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/03/664103/1/check/openstack-ansible-functional-ubuntu-xenial/7f281ff/ : SUCCESS in 23m 50s","accounts_in_message":[],"_revision_number":1},{"id":"55cedcbd304e62b7a802dd2d6f68cd3e9834fb1c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-27 15:32:22.000000000","message":"Abandoned\n\nos_searchlight repo is now retired","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4b9217a079f21d1723c4ec9daac0cde766d5f74","revisions":{"e4b9217a079f21d1723c4ec9daac0cde766d5f74":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:28:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/664103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_searchlight","ref":"refs/changes/03/664103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/03/664103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/03/664103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/03/664103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/03/664103/1"}}},"commit":{"parents":[{"commit":"0c7e1ada77e5d8d1744f5968ccec23a84c969649","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/0c7e1ada77e5d8d1744f5968ccec23a84c969649"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:28:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:28:20.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I50092ea057e74a694aaa29874abccc1423033209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/e4b9217a079f21d1723c4ec9daac0cde766d5f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/e4b9217a079f21d1723c4ec9daac0cde766d5f74"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
