)]}'
{"id":"openstack%2Fopenstack-ansible-os_searchlight~664343","triplet_id":"openstack%2Fopenstack-ansible-os_searchlight~stable%2Frocky~Iae3d4f66691f1de58a59148b6bfc89d5eb0199ad","project":"openstack/openstack-ansible-os_searchlight","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iae3d4f66691f1de58a59148b6bfc89d5eb0199ad","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:45:10.000000000","updated":"2019-06-10 18:05:27.000000000","submitted":"2019-06-10 18:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664343-1560189927677-8840775f","meta_rev_id":"b593144bc6a80fff7c8cfbee3903e3e19b8de226","_number":664343,"virtual_id_number":664343,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 18:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:07:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:07:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 18:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b61f85f815eaaa2b444c10dc69a9facfecd996b4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09715c8f1ca7c27cbf454378a4cbc96cc44cc803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/43/664343/1/check/openstack-ansible-linters/5bed161/ : SUCCESS in 6m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/43/664343/1/check/openstack-ansible-functional-centos-7/a790570/ : FAILURE in 31m 12s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/43/664343/1/check/openstack-ansible-functional-opensuse-423/0a58773/ : FAILURE in 12m 11s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/43/664343/1/check/openstack-ansible-functional-opensuse-150/99e6dc3/ : FAILURE in 8m 38s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/43/664343/1/check/openstack-ansible-functional-ubuntu-bionic/4487782/ : FAILURE in 30m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/43/664343/1/check/openstack-ansible-functional-ubuntu-xenial/d089580/ : FAILURE in 38m 05s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/43/664343/1/check/build-openstack-sphinx-docs/40848ad/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a7f6a845d3365edd89de6a3972208335b23e366b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:07:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7afd941895e9f195ade26b96491dcdd61dfd7c51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:07:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dadc549281400e70c367a5f911a4c16554c30ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f96ea843506535c30c5ab93ff68bf3ccfba625e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:05:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s\n- openstack-ansible-linters http://logs.openstack.org/43/664343/1/gate/openstack-ansible-linters/47ef94d/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/43/664343/1/gate/openstack-ansible-functional-centos-7/7b0493f/ : FAILURE in 45m 52s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/43/664343/1/gate/openstack-ansible-functional-opensuse-423/206749c/ : FAILURE in 7m 09s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/43/664343/1/gate/openstack-ansible-functional-ubuntu-bionic/f1d7387/ : FAILURE in 35m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/43/664343/1/gate/openstack-ansible-functional-ubuntu-xenial/6f06d17/ : FAILURE in 29m 31s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/43/664343/1/gate/build-openstack-sphinx-docs/753398c/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abe2a391209c66e61909a03326b2b03e067ad3b8","revisions":{"abe2a391209c66e61909a03326b2b03e067ad3b8":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/664343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_searchlight","ref":"refs/changes/43/664343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/43/664343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/43/664343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/43/664343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_searchlight refs/changes/43/664343/1"}}},"commit":{"parents":[{"commit":"e2265d62f394fe954393a2b076d01447aa3534a9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/e2265d62f394fe954393a2b076d01447aa3534a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:09.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iae3d4f66691f1de58a59148b6bfc89d5eb0199ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/abe2a391209c66e61909a03326b2b03e067ad3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_searchlight/commit/abe2a391209c66e61909a03326b2b03e067ad3b8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
