)]}'
{"id":"openstack%2Fopenstack-ansible-os_senlin~749365","triplet_id":"openstack%2Fopenstack-ansible-os_senlin~master~I3abd0d0099f82f57eea0cac1fdd9050592aefeff","project":"openstack/openstack-ansible-os_senlin","branch":"master","topic":"749365","hashtags":[],"change_id":"I3abd0d0099f82f57eea0cac1fdd9050592aefeff","subject":"Initial commit to os_senlin","status":"MERGED","created":"2020-09-01 17:12:32.000000000","updated":"2020-09-16 21:49:31.000000000","submitted":"2020-09-16 21:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749365-1600292152404-72df3f54","meta_rev_id":"3b14d32807cee4d79de8616abcdcd3e8b7a8d342","_number":749365,"virtual_id_number":749365,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 21:35:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"recommended":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"all":[{"value":1,"date":"2020-09-09 03:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":2,"date":"2020-09-15 13:09:07.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-09-08 20:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":1,"date":"2020-09-15 13:09:07.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-09-16 20:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-06 12:59:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-15 13:09:07.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-09-16 20:04:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-09-16 21:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e94a0a4cc9be054c309e1cb1b4629a6e1d707565","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-01 17:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b179df4ba502f007174cdbe7d23b6d4c489d93bb","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-01 17:47:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c64e1ec1f7d7e412950e04054095d7813abaf32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:02:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf89b577535e4a1a8e474a561a2db12e : SUCCESS in 2m 49s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/72f5c118760f425a80a62e6c29b5a032 : FAILURE in 8m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e12f1fce7b74030b9a2e52f5724534c : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8e278963bdd44f383405ea0ab23540c : SUCCESS in 1h 06m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46e48e54e53c4369ae3b3ef4fd5650ca : SUCCESS in 53m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f42b8534d0bd4d93bc7a409cbfe6d7f9 : SUCCESS in 58m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ab05398c20224c8c8bb63feb3dc388bb : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e474396f7d0940289ae09f95cd6db829 : FAILURE in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/153fa88842b14b87ad76864f2f8c7a33 : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b1c5f2e80fd4107886ea351d81d6413 : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fdb86c79a494233bbc526b624b1f2f0 : SUCCESS in 1h 03m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e4ded07efba4eb8b016bb907dfb1bfe : FAILURE in 2h 08m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/526a10875e18455a8cf46daed4574970 : SUCCESS in 1h 58m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7e6842755547c68727db25c374ccf5 : FAILURE in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/893efb8bc04b4f08be955610c941bb90 : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"68a0b8de3cfebb173470191bfc66332f654a712c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-02 15:14:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24fd6096f88940bdd38df97e704cfa76ba29d0b5","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-02 15:14:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3db7e8c0277566238ce34049ec067554b16e448e","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-02 15:53:47.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"dfee7d2a659d269e784b11f760ea77c8acba6195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:25:00.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"31c0d78ba663377fb608cbb866edf4d368c3fbde","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-02 16:25:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e32527b74410a9c493ab1ca34f76c51cea744c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:41:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/89861cd9e0c244e3aa6745e87901cdb4 : SUCCESS in 5m 36s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e8e9b677191840b28d1fa4eaa638a598 : FAILURE in 8m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1ee67b611a1e4f989d2110d66889dc9f : FAILURE in 48m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e8d88215d80a495296ae8f6453dbd6b0 : FAILURE in 46m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2ba0561c64d4e1ebc94e21f39714b05 : FAILURE in 49m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00236060286642babff6ddcdc2a3d75e : FAILURE in 43m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c3aaf3361eb54338ba2a7eea57f3facc : FAILURE in 52m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1ca277d03b1f43559525bbb3d955d043 : FAILURE in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/da7302e84eb4438aa29f9573cf4f733f : FAILURE in 29m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35d565d7b2c440d6b8511e3cd1a54fb3 : FAILURE in 57m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e380f7ff1c34b4a9fbb8de1d77045b7 : FAILURE in 1h 02m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7add63dbf9d24c878e9d5d36e6ed4cfa : FAILURE in 2h 05m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb6fdcd41c54438fa415ebb785dcfa23 : SUCCESS in 1h 48m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2772aaaabdbe49fab7c37952c4c50bdf : FAILURE in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/319dcb9b7f4742cfad23ff8a41d37625 : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":4},{"id":"db86d7beee3189bfda8c34a96c410df116e1c025","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-02 19:15:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bd3ab9f116c5ad7552b12f766d8e8e07739e7397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:48:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea84301543f346d28883763a1fd6e56e : SUCCESS in 3m 07s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6298d71237ed4f828b71ca262c15ebc2 : FAILURE in 8m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2b68b3b9931043f791b69526725fdb36 : FAILURE in 51m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e2a30725d8b14536b1e55921a7dcbf5f : FAILURE in 47m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/028749570fc141678ac7214c3796b7d7 : FAILURE in 25m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e39f21dcb8f14c42b2fb82a2c05f4e5e : FAILURE in 33m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f1f9ad0bb0774fab90151d22030ac0c1 : FAILURE in 39m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d72100cb9504924ac550c8d5855145b : FAILURE in 1h 09m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e9bf2f6a57f4063855767d3a104dcfa : FAILURE in 26m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee45bcf90a8d4af798d03777be3826e5 : FAILURE in 22m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fec0f144aff46e286a2e1a235b981de : FAILURE in 52m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c91117f62124c8ba3b50fd711ceef4e : FAILURE in 2h 12m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5ace600e5104c25aeca96ad430bf571 : FAILURE in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a220b98c0484923a715b99d9d7fc18b : FAILURE in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38923826589d4564a250a91a982efd16 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a355e95488d38b41068a3c4ab068d8bdb20a5180","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-03 06:00:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3967033e95f35221167390698bc1ea5714ecfadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 08:36:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e33c0d457b0a48108c726a3905807703 : SUCCESS in 2m 49s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0341645e35bd4e85a63d35ae9c68ca67 : FAILURE in 7m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9db2acdf49da45e3a96da99ea0fec55d : FAILURE in 53m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e4bcf17da3e245c1aedd618a2b028951 : FAILURE in 53m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc3643a0d601457cbf11871c853e3c60 : FAILURE in 55m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147a95be471641a989513b140d0b3bc2 : FAILURE in 47m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/52ac8eff68ca4999885bb2cbce2c91eb : FAILURE in 59m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c258445227a1475e87f224c36ea84b86 : FAILURE in 53m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c0671f16a753490d8ced21b2127963b2 : FAILURE in 53m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e1f6599d6a04746a47ca23683c80a89 : FAILURE in 58m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6190caa349e84211840aa65df94cd7c0 : FAILURE in 50m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2475b000f9d1424797a411783170a18b : FAILURE in 2h 05m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/545e309cacd14c23b3da14495b177e62 : SUCCESS in 1h 40m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3803aac3f854b4bb6b8369cce18f660 : FAILURE in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0ac4e3e1d784fea841311ffcc37e6fe : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":5},{"id":"0274d447ebeb3ab49640df6350e624f17b5ae2f2","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-03 13:09:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b06143ba5a7d2bc90ddb1020df9464a787660ea6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-03 14:59:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"17fc284611a533de2a044938ec336af487cb6b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:17:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/55c4533ba1304602be9be8174bc1e2d7 : SUCCESS in 2m 37s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3b8a2453ddc479c9f925a48a52e42b9 : FAILURE in 7m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/489c5d38fbc74ed380ce3d3d21172557 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec91ddf3e8914947bc19d53d3803bb73 : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c25917f9212b4ef79041ea9a7f3c17df : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4f8f625ed1346cfbe00b34bac86638d : SUCCESS in 1h 43m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/81a26f9f7a7247b2bb2cdf79601c8278 : SUCCESS in 1h 09m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a2b74c418c9c41988feaca803d5073fd : FAILURE in 1h 50m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/052012d2c2f44e69980c5fe9d2c2b6ef : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c93fbc51f8149fab93b61af505bcbed : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0137a8b602644fac831d6c6c59da34fe : SUCCESS in 1h 11m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cb89e0bff59a4e63a261eabe86d8ee53 : FAILURE in 1h 59m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b08edd7e65cf472397aa9dc80a433760 : SUCCESS in 1h 46m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e82aaac40c4c659767271b5bc20e15 : FAILURE in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95f0d006f71f42b1a542d147db7c488c : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":6},{"id":"d325acab6eb3934ebc6dcb463e3c35e0b150bcf3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-03 22:20:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"73a67dd0d1ceb78f97dd06964a9a865b3920953d","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-04 14:21:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c9d2c5c7df8b8cbd726ef7b8ec05751608be01d5","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-04 14:22:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d1ee1a6871c7b4e64b6363be1e9c95f4a6940907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:45:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/54ee66c2a4d94415a308b2e75a9657ef : SUCCESS in 3m 15s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2c36c4ace35d4e00b82712e8baa9e404 : FAILURE in 11m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ba341f5a06be4c879eb5aa9dfbe52cb5 : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6cc93f2bbd46432cbc890142bb8ab681 : SUCCESS in 1h 01m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c9b4f46cc764e50a67b75e8900776dc : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf0403bbbcf416a8004e0e961f7dd22 : SUCCESS in 59m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/91b309783d404aa99847f261e0ce674e : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c6d21be821643409eb10c7e86bd95de : FAILURE in 1h 27m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ca975756f4a842888f8d2f29d0d4b808 : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58333982cefb426081ed7fb5f19de1e6 : SUCCESS in 1h 30m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e938b7762f345cea79ef833625e8bc7 : SUCCESS in 1h 17m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/81c14eea8d7b400fac096ee34854aa76 : FAILURE in 2h 13m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0519690794894a11ab47be662a6b9b1d : SUCCESS in 2h 05m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14381a0426dd499d90358d58c9c467a6 : FAILURE in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5159c1708d40afae5792980cef9e3d : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"ac7e952b5b8b116a121df8d1c8d55bc6f5e1cbd7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-06 12:59:04.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"704c308a77a093f38e324a6d39b325c5805b29d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 15:37:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/98204883dd184981bc1731d647151e74 : SUCCESS in 2m 48s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6dbcb5489f704f599583e9c44295c2e0 : FAILURE in 7m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4bc09369e9e141fb98fca49d414dbefa : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c1484be5d054cb1a96656b04c979650 : SUCCESS in 59m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e5db3098824e4a63a153f30744ab5c9a : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/84d0361ead5c42b0ac0f64c7ff9faa79 : SUCCESS in 2h 00m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/610c8d5572904418b596ab0628ec0b91 : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28c82c24cfbf47d6a5172c6113ec00f1 : SUCCESS in 1h 51m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12f2c68ed9cf49c9a2109cb80887675c : SUCCESS in 2h 05m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/645991de40014c74ba05d39dcc461de2 : SUCCESS in 2h 34m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151ee2e3e5c140f79a5cae592225676a : FAILURE in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2782e8305fab43c1ade68ee9ed2152d8 : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":7},{"id":"4f02af5796c289da6db8d8f68faed598570010c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-07 08:48:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"14170bc82128cb4294ffe42ac9b0fc12f9ff2489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 11:40:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6db9fcc59c8b41fcb16245507a652687 : SUCCESS in 2m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/942776d452104efb874677265ce17652 : FAILURE in 7m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e55d3c7943d546269163671312a2a71c : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd697c2f1d094cb698fb71d9513b680d : SUCCESS in 1h 50m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d9cb9874ef4b48209154140026173c62 : SUCCESS in 2h 04m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6440acea47ff4263a9ea4597acf625ff : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c18af7f4f1aa48a99a56ca400aa4af96 : SUCCESS in 2h 46m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/941a233ac06d4cf0afd25bc5efb70f9a : SUCCESS in 1h 23m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dff335422c144f7dbc53a2a38917428d : SUCCESS in 2h 22m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/611afb2a7f694ba2bcb3ce0f55894029 : SUCCESS in 1h 57m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8ebf828e18471ab63e6515f853e289 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa99499252b4355acec9251ef87ff50 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":8},{"id":"62fe077d5e4f5882cef9f43e3715dc6d8f50fa34","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-08 08:12:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"aea16d1ca355da522db45f78d17491c36e27c6b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 10:21:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dcb19ac7d09e439fb216d22bbf7670e9 : SUCCESS in 2m 43s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a4a85a3eb214ab1a86dc821687959ca : SUCCESS in 15m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9aee7d4fe0694c3ea2fb2e4910b180c0 : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7746b38d59754564b2a1915111ecc842 : SUCCESS in 55m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6d8e0caf303c4a8aa987deffbad2f361 : SUCCESS in 50m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9f85d483b63348b7893e0fd1f42b5992 : FAILURE in 32m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c131d84d3a1494dbd1fd1ae91554212 : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/200055f8effa4ee2a614686a5e74b9a2 : SUCCESS in 54m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2444519a86264c818c9ba785cfa1619f : SUCCESS in 2h 01m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a690666ddec94681bf0b76a884a69a93 : SUCCESS in 1h 07m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8ae6681627425fa996246f3661a6db : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/598cded693304326bc8cf233a2c5a304 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":9},{"id":"f0a6639d378ed5f799098d8d3a051062dbb5db73","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-08 14:52:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2f058fff3cf7795b895c9207c3c6d452049bf3be","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-08 20:30:23.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"7e535ffb87953d98b90a87e8f88adad9fd143830","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-09-09 03:30:27.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"53ff9a97fcc455c67d9299bae27b57b384e1a505","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-09 06:03:51.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"adedb7832ae982df09b242e510c38c26b99df3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 08:25:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e5f77c1de8040269a5973c896d853a7 : SUCCESS in 2m 40s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d605596322db4e6d9d1ec1f0a2ea6cae : SUCCESS in 20m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a32dfbc16bb4d2fb7cdfbc633819446 : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/911975166ccb4e2f849e4237b1a4596d : SUCCESS in 58m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7455b64ffc5240369a851dd41a6e88dd : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e93031df5c7e4410a43b66023bb55347 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccf19e713fa14289ba8a7e50769e3861 : SUCCESS in 1h 06m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c3cafe0a75a4998aa82039e9390aab1 : SUCCESS in 1h 00m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1f60512b62b240f4811cd12e74a6d474 : SUCCESS in 1h 48m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b022b6c3de5048d6b4993d31ca514381 : SUCCESS in 1h 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e108986b7ce45cbbdf638aa175aae68 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b6d0e9bb9145b6a72d9d6efffe6dc2 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":9},{"id":"48ef83a5d30f45010e7eebe11f059cde2fbb9e2d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-09-15 13:09:07.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6d2612092a964ceb0fd31e9a7ea12a631e5ebd4a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 20:04:08.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8d223f5eff2d024dc83a6d9425d5a333d9cf628d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:04:23.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"63f3f75194072850c7b97799afa20004b1f35ff6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:35:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"10156c60fa39090425460f2472ce792be0643149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:35:52.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c2edfc73e084cbf8160aac85f802450 : SUCCESS in 3m 02s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e57fa2ecf54b413c9dd0ddaedaba7732 : SUCCESS in 14m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f42ec2c887004989bbc290e8536d7bfa : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/153fc12a0aff473aa43659794949354a : SUCCESS in 54m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/878aee46824c4fa299ca4f3a6a55b8b5 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/06548291b503470a8140e283705ecfa8 : SUCCESS in 1h 25m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/798b8e8d1c184be794ad6d6659623d17 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/329bc1fd81c14e5ca633a6b48e26b029 : SUCCESS in 1h 03m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab288d316ac14772b1884311821d3c2d : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcba7858923144bab6d6261fc2eaf262 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":9},{"id":"c478dab51f67e5515297cb1a7528e84ae0e23e93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:49:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f41adfe30ef488aa141c951125f7896 : SUCCESS in 3m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d40162b5de3f4df8a8dc741af1cba57b : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b8ee7bb2b1689c7422d690d54e5ce837b6326ff6","revisions":{"a9383aa482ec9bcd150f948e283f1961163e4059":{"kind":"REWORK","_number":1,"created":"2020-09-01 17:12:32.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/1"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:31.000000000","tz":-240},"subject":"First commit of os_senlin role","message":"First commit of os_senlin role\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/a9383aa482ec9bcd150f948e283f1961163e4059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/a9383aa482ec9bcd150f948e283f1961163e4059"}]},"branch":"refs/heads/master"},"f94b594f98570670c8694b955d061b88eb736ba3":{"kind":"REWORK","_number":2,"created":"2020-09-01 17:47:31.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/2"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:45:04.000000000","tz":-240},"subject":"First commit of os_senlin role","message":"First commit of os_senlin role\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/f94b594f98570670c8694b955d061b88eb736ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/f94b594f98570670c8694b955d061b88eb736ba3"}]},"branch":"refs/heads/master"},"e23433a23c10eb03d01570f740a98a113fbb81a6":{"kind":"REWORK","_number":3,"created":"2020-09-02 15:14:42.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/3"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-02 15:13:27.000000000","tz":-240},"subject":"Fixing some defauls/mail.yml tunable options","message":"Fixing some defauls/mail.yml tunable options\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e23433a23c10eb03d01570f740a98a113fbb81a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e23433a23c10eb03d01570f740a98a113fbb81a6"}]},"branch":"refs/heads/master"},"4f77b1870d03cdb8523aca33b93ff2fc6b283ced":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-02 15:53:47.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/4"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-02 15:51:16.000000000","tz":-240},"subject":"added os_senlin role for deployment.","message":"added os_senlin role for deployment.\n\nDepends-On: https://review.opendev.org/749540\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/4f77b1870d03cdb8523aca33b93ff2fc6b283ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/4f77b1870d03cdb8523aca33b93ff2fc6b283ced"}]},"branch":"refs/heads/master"},"c9eccff2d8672d5d3bce8e2da7dbe060c3ac95f7":{"kind":"REWORK","_number":5,"created":"2020-09-03 06:00:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/749365/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/5"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-03 06:00:26.000000000","tz":0},"subject":"added os_senlin role for deployment.","message":"added os_senlin role for deployment.\n\nDepends-On: https://review.opendev.org/749540\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/c9eccff2d8672d5d3bce8e2da7dbe060c3ac95f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/c9eccff2d8672d5d3bce8e2da7dbe060c3ac95f7"}]},"branch":"refs/heads/master"},"7e0dbec0adcb7defcd2fba3c59389f8c71c3d6d1":{"kind":"REWORK","_number":6,"created":"2020-09-03 13:09:26.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/6"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-03 13:07:37.000000000","tz":-240},"subject":"Adjusted senlin_pip_packages: and changed senlin_container_mysql_password to senlin_galera_password","message":"Adjusted senlin_pip_packages: and changed senlin_container_mysql_password\nto senlin_galera_password\n\nDepends-On: https://review.opendev.org/749540\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/7e0dbec0adcb7defcd2fba3c59389f8c71c3d6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/7e0dbec0adcb7defcd2fba3c59389f8c71c3d6d1"}]},"branch":"refs/heads/master"},"8ea46ebe1664e721c911bb3262c2fca258c295a3":{"kind":"REWORK","_number":7,"created":"2020-09-04 14:21:09.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/65/749365/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/7"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-04 14:19:48.000000000","tz":-240},"subject":"Adjusted senlin_pip_packages: and changed senlin_container_mysql_password to senlin_galera_password","message":"Adjusted senlin_pip_packages: and changed senlin_container_mysql_password\nto senlin_galera_password\n\nfixing auth_url in templates/senlin.conf.j2\n\nDepends-On: https://review.opendev.org/749540\n\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/8ea46ebe1664e721c911bb3262c2fca258c295a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/8ea46ebe1664e721c911bb3262c2fca258c295a3"}]},"branch":"refs/heads/master"},"cb9f9e38a07b43c8fd6b5f29548994c6ca06aa89":{"kind":"REWORK","_number":8,"created":"2020-09-07 08:48:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/749365/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/8"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-07 08:48:38.000000000","tz":180},"subject":"Initial commit to os_senlin","message":"Initial commit to os_senlin\n\nDepends-On: https://review.opendev.org/749540\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/cb9f9e38a07b43c8fd6b5f29548994c6ca06aa89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/cb9f9e38a07b43c8fd6b5f29548994c6ca06aa89"}]},"branch":"refs/heads/master"},"b8ee7bb2b1689c7422d690d54e5ce837b6326ff6":{"kind":"REWORK","_number":9,"created":"2020-09-08 08:12:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/749365/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/65/749365/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/65/749365/9"}}},"commit":{"parents":[{"commit":"fd0afb07a08c3d9c777ab6ace7271958e2e63f7a","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/fd0afb07a08c3d9c777ab6ace7271958e2e63f7a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-09-01 17:12:03.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 08:12:31.000000000","tz":180},"subject":"Initial commit to os_senlin","message":"Initial commit to os_senlin\n\nDepends-On: https://review.opendev.org/749540\nChange-Id: I3abd0d0099f82f57eea0cac1fdd9050592aefeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b8ee7bb2b1689c7422d690d54e5ce837b6326ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b8ee7bb2b1689c7422d690d54e5ce837b6326ff6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
