)]}'
{"id":"openstack%2Fopenstack-ansible-os_senlin~764650","triplet_id":"openstack%2Fopenstack-ansible-os_senlin~master~Ie68fae480f1ab1450659e1b4b2f75f65360fce7f","project":"openstack/openstack-ansible-os_senlin","branch":"master","topic":"api_threads","hashtags":[],"change_id":"Ie68fae480f1ab1450659e1b4b2f75f65360fce7f","subject":"Reduce number of processes on small systems","status":"MERGED","created":"2020-11-30 11:57:52.000000000","updated":"2020-12-01 16:13:39.000000000","submitted":"2020-12-01 16:09:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764650-api_threads","meta_rev_id":"5a75b8505dca3052ecc85bc7cc35c8d732097411","_number":764650,"virtual_id_number":764650,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 16:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"all":[{"value":1,"date":"2020-12-01 02:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":2,"date":"2020-12-01 14:07:01.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-30 14:15:36.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":1,"date":"2020-12-01 14:07:01.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 14:00:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 14:15:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-12-01 02:22:03.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2020-12-01 14:07:01.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"db3754e00b480c8600e823847b044b846067a81f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-30 11:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e3df35fbede43fd0d4d1ce42759994a360a73a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:00:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2c689138e5147c3bb804b08bbc462d3 : SUCCESS in 14m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b8d5b8fcfed64b8fb5a140ebb55f76eb : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67745832c07c4c928f5400a6217de334 : SUCCESS in 58m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e5837c7fedaa477495e1a5f7ec88bcd2 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4393e49bdba9461faeb2db0eedc4f92a : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be1f1bb8ea47482a8d1816d9d311f80f : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baf1622259794c02a482c444d1719e86 : SUCCESS in 1h 13m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/329af8e2f54040c09cdbca248a286158 : SUCCESS in 1h 53m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b39b4d104a7e4ae8ba6c9a565e4a7105 : SUCCESS in 1h 36m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f8116f5fe54e51ae9628db979f9819 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"36d0e645a22dfd4fdf34a3e1575c06646b397d18","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-30 14:15:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b56701fca9341a74aaeed24ce16175f422da2c4","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-12-01 02:22:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d79d43b35296434b12c38b5c6a6f34da23b7434f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-12-01 14:07:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"030d416b34fddf45519d706ae2b6ade98870cf5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:19:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fde3b84481ad1f4eab62b25dc0719cb2d0ab1da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:09:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a061c55f1c847ceb954448691b66c84 : SUCCESS in 15m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b5df1246e3d645b6a2f9eaf278c57090 : SUCCESS in 1h 05m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf794efc7b004a388e7d4fbeb04654f2 : SUCCESS in 58m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b978643a510847348cd749090dad38ee : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/580bfec0de9b430ea7a6b878857e6c41 : SUCCESS in 1h 36m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/877602f4391e4e9e86110f02ae72a5b4 : SUCCESS in 1h 28m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe5aa98443ee4f5fb011ed1ceacee048 : SUCCESS in 1h 15m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb33031af3844ccc84a1031b16c07801 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1883a70d7845f2c455109202317fa96da5691843","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:09:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a75b8505dca3052ecc85bc7cc35c8d732097411","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:13:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e212b1babccd4622a560ec426344ddf0 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0eb41f868f6513ab55514f4bccb0a486518ef0f","revisions":{"b0eb41f868f6513ab55514f4bccb0a486518ef0f":{"kind":"REWORK","_number":1,"created":"2020-11-30 11:57:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/764650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/50/764650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/50/764650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/50/764650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/50/764650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/50/764650/1"}}},"commit":{"parents":[{"commit":"120dac7ef6627d9cb4f127a2584e674e737fe72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/120dac7ef6627d9cb4f127a2584e674e737fe72b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-30 11:57:43.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-30 11:57:43.000000000","tz":120},"subject":"Reduce number of processes on small systems","message":"Reduce number of processes on small systems\n\nEven the most modest 4C/8T system would run with the maximum 16 processes\ndue to the calculation being VCPU*2.\n\nWe devide amount of CPUs to number of threads for hyperthreaded CPUs\n\nChange-Id: Ie68fae480f1ab1450659e1b4b2f75f65360fce7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b0eb41f868f6513ab55514f4bccb0a486518ef0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b0eb41f868f6513ab55514f4bccb0a486518ef0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
