)]}'
{"id":"openstack%2Fopenstack-ansible-os_senlin~777296","triplet_id":"openstack%2Fopenstack-ansible-os_senlin~master~Idc41208fdf182e4ffd7b1b7d4940034697b60904","project":"openstack/openstack-ansible-os_senlin","branch":"master","topic":"osa/senlin","hashtags":[],"change_id":"Idc41208fdf182e4ffd7b1b7d4940034697b60904","subject":"Allow Enable/Disable feature of uWSGI support","status":"MERGED","created":"2021-02-24 04:55:33.000000000","updated":"2021-03-25 17:58:14.000000000","submitted":"2021-03-25 17:55:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777296-osa/senlin","meta_rev_id":"2b308b286a817b3649597a30ad3b2d99da03ffc2","_number":777296,"virtual_id_number":777296,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 17:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"all":[{"value":1,"date":"2021-03-24 14:39:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":2,"date":"2021-03-24 14:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-25 10:47:19.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-25 10:47:19.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 06:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 05:18:54.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-03-12 05:18:54.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-24 14:39:39.000000000","updated_by":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"reviewer":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"state":"REVIEWER"}],"messages":[{"id":"53fb0fd34fb46885d56955f7969274a3a5a86d1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-02-24 04:55:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3b246fbfe5f7de171e1f53edd33d1f938654638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 06:49:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace64bc9335346e080ff693a24e6b344 : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8e453649fab7469da71aee16d9c7f8f6 : FAILURE in 32m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6889c0294eb440edaf3a5c293455dcf3 : FAILURE in 38m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/287392f3c7e742ef8e3f762f757e1eb7 : SUCCESS in 59m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d19a558eaf54ec9b64b5658890b1e0e : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f105b9cfbe474099bb1f78d0df126846 : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1017090a8d2149368aba1c7b6f3613b9 : SUCCESS in 1h 02m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/04d847fd31f14c348744a2c6a60755a5 : SUCCESS in 1h 51m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00fedf3d8a4041108d69150a4567caac : SUCCESS in 1h 36m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aca7276dc37499687ece20905c9a21d : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d2326a15acd294b56d0b4d16965e7dd4a2d05d64","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-02-24 13:41:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcf79b8da08adf1c93cb50d83ce1b14176c40d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:20:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c0673d810d48daaaae2d4a9ca24770 : SUCCESS in 15m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2da5f07eed504463889f01acc278c47e : FAILURE in 50m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfface82733f46118c31ce3247ef59c0 : FAILURE in 48m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/607b593c45ce4884b81c2c305733e653 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/be52d13654004d93bd8cd8b934f4977a : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76da7e2ceae3415bbd454efe43050957 : SUCCESS in 1h 49m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f631b3b4aa3f49e4b3d177751d1696f8 : SUCCESS in 1h 22m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/33fbb99d238e4c678a70b47b2f6e286e : SUCCESS in 2h 13m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba0d290f800e40f2aa9e572dfc888488 : SUCCESS in 1h 44m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0990abf2c1a4ed2bc56b62765a42ada : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":1},{"id":"494ec3ac1757c08d76a22594fc37e8989edc230f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-22 12:47:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c79b6e539e6d710b58fb194bf44d435522b02fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:43:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b673218957f44f0484d0f45837281ffd : SUCCESS in 19m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/950f6512230b4bdfbbf5f49687d4a08f : FAILURE in 38m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d3a55ef8ac1433ea55de3137e299de4 : FAILURE in 50m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cd6944b3c4374d849243a4d158151794 : SUCCESS in 1h 15m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4939aa3a6be6420a9a8421885bdf2b6a : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcb961fe3eaf403d81083e40caf9f77e : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6303f558cb5548c1af44e46af795ce08 : SUCCESS in 1h 17m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/100341587c3c4d9981888b1e7d7b5020 : SUCCESS in 2h 00m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740627b24e70486b8e83d6628f9679ae : SUCCESS in 1h 47m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0093ccd208d44a4bbbfbbcbad1710652 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"452f9ec242893d97cd9c677b24393a4a767cb6af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-23 18:12:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b676ccdcb5be9c42ea215088e20a9e5cacba35cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 20:08:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17ce28f8af8948299ac7571de5562bfc : SUCCESS in 16m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e562ae1eef9042f680f2e4d9ba43bbaa : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed005a327b74ed998b0d67775c62920 : SUCCESS in 44m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e68c60c7de6b44c39c17f63de6391dd0 : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2734b4e26d644d47aa9a207c6de75a2e : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59a252fb309146dc8a32ff97d1563b2b : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eb058f3b4504906924e38b4e1caec4d : SUCCESS in 1h 15m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b4aa9cb9d17c4390897b4c6702ab589f : SUCCESS in 1h 49m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cfaadcc3c714c84b049c967d7ec82e6 : SUCCESS in 1h 39m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0dc09b68fbf4c37ac9ea120be69d619 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"78e09f406363ad3f78e4ac4484fd1b2e82efe75c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-24 14:24:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4b6f9ca0daaef2f6e58c29399761c8ebbdd371aa","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2021-03-24 14:39:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6d706ad3af71297dc0321455ad26371faee0c796","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-25 10:47:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ca722f872534b9fe04d06126852a116965dfbdbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:26:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59eb9b9847987f372085a8868f97b33e627a4b08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:55:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/908195f3cf7f4c79be60d5b3be08fdf3 : SUCCESS in 18m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/218d687a17ec4e88a1f5947d9bbc4cce : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/223ee5c0d4fc480782650cc27535962b : SUCCESS in 1h 01m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/266421e4cd91493d83d1b08f64cd8b66 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ec41837469440f5b5b01ff52c6c5343 : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9b1e5163c504134a88c4104974fd10e : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d727d9afbd243469a09e52a2f862754 : SUCCESS in 1h 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6826c793e7d64c94bf3c08151f22afe6 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":3},{"id":"315b70da33f1e845c651cc11690f82e8336245dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:55:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2b308b286a817b3649597a30ad3b2d99da03ffc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:58:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28801ca50b4349c0ae72ded6c69d5332 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57276fd0abad3c7c8c79de51573db7956e604d9c","revisions":{"b8421b347d7aa57568baf7cdfaaf7cb190f43264":{"kind":"REWORK","_number":1,"created":"2021-02-24 04:55:33.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/96/777296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/96/777296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/1"}}},"commit":{"parents":[{"commit":"62df77bfaaab2e212be202ca14fe84ceb2e6a8dc","subject":"Merge \"Use global service variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/62df77bfaaab2e212be202ca14fe84ceb2e6a8dc"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-24 04:52:29.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-24 04:52:29.000000000","tz":0},"subject":"Allow Enable/Disable feature of uWSGI support","message":"Allow Enable/Disable feature of uWSGI support\n\nNoticed uWSGI causing strange connection reset issue between senlin-api\nand rabbitMQ, so adding support of variable to disable/enable uWSGI feature.\n\nChange-Id: Idc41208fdf182e4ffd7b1b7d4940034697b60904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b8421b347d7aa57568baf7cdfaaf7cb190f43264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/b8421b347d7aa57568baf7cdfaaf7cb190f43264"}]},"branch":"refs/heads/master"},"766533b57302ff5db825076965bf2209be89efb0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-22 12:47:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/96/777296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/96/777296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/2"}}},"commit":{"parents":[{"commit":"57cbea0130bc31bb405a7f682b8597a984502115","subject":"Correct python egg name for senlin source install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/57cbea0130bc31bb405a7f682b8597a984502115"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-24 04:52:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-22 12:47:00.000000000","tz":0},"subject":"Allow Enable/Disable feature of uWSGI support","message":"Allow Enable/Disable feature of uWSGI support\n\nNoticed uWSGI causing strange connection reset issue between senlin-api\nand rabbitMQ, so adding support of variable to disable/enable uWSGI feature.\n\nChange-Id: Idc41208fdf182e4ffd7b1b7d4940034697b60904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/766533b57302ff5db825076965bf2209be89efb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/766533b57302ff5db825076965bf2209be89efb0"}]},"branch":"refs/heads/master","description":"Rebase"},"57276fd0abad3c7c8c79de51573db7956e604d9c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-23 18:12:03.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/96/777296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/96/777296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/96/777296/3"}}},"commit":{"parents":[{"commit":"adb72c8d8a5e3af48911243f5f6b386d8e628e46","subject":"Correct python egg name for senlin source install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/adb72c8d8a5e3af48911243f5f6b386d8e628e46"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-24 04:52:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-23 18:12:03.000000000","tz":0},"subject":"Allow Enable/Disable feature of uWSGI support","message":"Allow Enable/Disable feature of uWSGI support\n\nNoticed uWSGI causing strange connection reset issue between senlin-api\nand rabbitMQ, so adding support of variable to disable/enable uWSGI feature.\n\nChange-Id: Idc41208fdf182e4ffd7b1b7d4940034697b60904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/57276fd0abad3c7c8c79de51573db7956e604d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/57276fd0abad3c7c8c79de51573db7956e604d9c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
