)]}'
{"id":"openstack%2Fopenstack-ansible-os_senlin~820241","triplet_id":"openstack%2Fopenstack-ansible-os_senlin~master~Idad975d6ee80c286f59ca69a8f99904bdf7939c9","project":"openstack/openstack-ansible-os_senlin","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-03 15:34:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idad975d6ee80c286f59ca69a8f99904bdf7939c9","subject":"Database connection pooling improvements","status":"MERGED","created":"2021-12-02 18:21:29.000000000","updated":"2021-12-03 15:41:04.000000000","submitted":"2021-12-03 15:34:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820241-db-pooling","meta_rev_id":"1bbecd2359d441613fdaf3b65bd26f5d534317fc","_number":820241,"virtual_id_number":820241,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 15:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-03 13:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-03 13:55:41.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 13:55:41.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 13:53:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-03 13:55:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9d4cc97d296a1d373830eecf82548234d578ec51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 18:21:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd101ce622b2226696babe1ebedc3742583881dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:48:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/401be4fec13e4a82a3efd9dcfc1dbb86 : SUCCESS in 15m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae007f99ba7a423d9cb2dd9d972acd5d : FAILURE in 42m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4361dd622d9478e9d7b0fa82fa301c1 : FAILURE in 45m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f8e3fe342c6d4142ad8f7481ad969a1a : FAILURE in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/af46e9b891fd42bcbc90e81ddd8df59e : FAILURE in 1h 02m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/315ff9bf05814fab8d482c45525dc031 : FAILURE in 1h 13m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a64bf2b27344b8b9aa735c6b6b2978c : FAILURE in 1h 17m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/643d33b768f64f3ca7842d4ad472d783 : FAILURE in 1h 22m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/915bd55cfd654cfaa8f26c922e297f21 : FAILURE in 1h 10m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/db16c8ab021449fb99eb7dd47fcee0bc : FAILURE in 2h 01m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c125198000e4983a47a04f730e702d4 : FAILURE in 1h 45m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bccca6b4e8e45eda08fe6210e4a9fc8 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a58a08cf319008291a0d5fe4498d35edb73e0014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 10:41:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1fa6a4c1ea32d229898356feda4cfce03d87b136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:06:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37cd7eb037644ff99098c0fc26a7e43e : SUCCESS in 17m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4e57a52e08a4e48ba522943c7f000b1 : FAILURE in 44m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f81cfdf2204457a64b13bfb58e40bd : SUCCESS in 53m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1622b8059b7c4b10821ae1e5721aec7c : SUCCESS in 1h 39m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d7c073f5e4a474db8147c77f6954db3 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8469b54fb9b44ddfa285d639f96d5019 : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6449521c33245e389a59c87281a0465 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f3875705e1a4b5996275e60a28a245d : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159a6719e0594d6c8e94968733c347bc : SUCCESS in 1h 36m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/06e32ba9c03e44cf9075b518abd5b6f7 : SUCCESS in 2h 02m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f460d61b25a04d639fd90c99edfcd176 : SUCCESS in 2h 02m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d5bddd9c9d4429bd6b22a7ec51fae5 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d7be2abbc466178981acb9753487b8fa662587f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 13:53:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bc08bdfb0f5b61b1f72e5ff16ade275ebf8c597c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-03 13:55:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8678ed3ab8cb954d2a323f39ea38bb6bbe3241f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:57:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a58cbe18cc06b2f61095b48fb3c2e72118962e29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:34:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/809e2b5c0cf94ef0a17d298a583dc643 : SUCCESS in 14m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c57ba993f214e4cbd5958b1899032f0 : SUCCESS in 55m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1976c723647149cd90044865ceece137 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2a1433b5dffb480a82b4f34dffed8aae : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea1ba188089e46a1901f67141d656439 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/45684614267146cab2956d18ee8925c2 : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb3e079a77914b6485fe06c69037288f : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccba6b44c71e4dc3b252afd3c9bd4347 : SUCCESS in 1h 18m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446bca6a6a4b413592f44b9e6497d830 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"049181b272b581fb605d80956a9307d69e2edac7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:34:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1bbecd2359d441613fdaf3b65bd26f5d534317fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b5966665b2474393263f3368d04415 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e0771d4e2c67e939a14711648cace0f4250d342","revisions":{"27667b6f2a37fa4ad56339e3cb528d37ebdc3657":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:21:29.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/41/820241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/41/820241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/1"}}},"commit":{"parents":[{"commit":"e4be01a8a3346f62787b570a65a26b0c95fd7fb3","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e4be01a8a3346f62787b570a65a26b0c95fd7fb3"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:21:20.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:21:27.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Idad975d6ee80c286f59ca69a8f99904bdf7939c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/27667b6f2a37fa4ad56339e3cb528d37ebdc3657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/27667b6f2a37fa4ad56339e3cb528d37ebdc3657"}]},"branch":"refs/heads/master"},"0e0771d4e2c67e939a14711648cace0f4250d342":{"kind":"REWORK","_number":2,"created":"2021-12-03 10:41:48.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/41/820241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/41/820241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/41/820241/2"}}},"commit":{"parents":[{"commit":"e4be01a8a3346f62787b570a65a26b0c95fd7fb3","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e4be01a8a3346f62787b570a65a26b0c95fd7fb3"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:21:20.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 10:41:44.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: Idad975d6ee80c286f59ca69a8f99904bdf7939c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0e0771d4e2c67e939a14711648cace0f4250d342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0e0771d4e2c67e939a14711648cace0f4250d342"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
