)]}'
{"id":"openstack%2Fopenstack-ansible-os_senlin~880554","triplet_id":"openstack%2Fopenstack-ansible-os_senlin~master~I7e0910f163560ce7b06c51264edc05ba43131522","project":"openstack/openstack-ansible-os_senlin","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-05-03 14:17:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7e0910f163560ce7b06c51264edc05ba43131522","subject":"Add TLS support to senlin backends","status":"MERGED","created":"2023-04-14 22:05:43.000000000","updated":"2023-05-03 14:19:08.000000000","submitted":"2023-05-03 14:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880554-tls-backend","meta_rev_id":"b6333dae1b07518774c2ec29a57893b710af29b2","_number":880554,"virtual_id_number":880554,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 14:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-03 11:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 12:22:24.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-03 12:22:24.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-15 00:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-03 11:00:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-05-03 12:22:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e87bf429b8c701d191d05b1e8d47f1af8238badc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-14 22:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40be36f2224c93a70e7ca8f28c5e5090914a1c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 00:44:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd2320408d54aeeb22b55bc3bfc733e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c84a5ef72a294d6ca5cd7a3c1f0596a2 : SUCCESS in 11m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2e626b195a34b09bcedaa7e0a885851 : FAILURE in 31m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adccecb89dfa4bfe88babd4b3acdaba1 : FAILURE in 30m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7e78be3f5dae4eb4b2cc47ada82b78a9 : SUCCESS in 1h 35m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/85a4804a293d4b599b29a26de094e017 : SUCCESS in 1h 22m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/321b1c0329ae4affb82edac6c9a9d921 : SUCCESS in 1h 47m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4086b9c8bd41e6b09816be2def819d : SUCCESS in 1h 45m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb5f3e4d705c472d826903d17f1a9c8c : SUCCESS in 1h 35m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d039441d3ed4b3ea20e8cb2b324826f : SUCCESS in 2h 29m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cff1c41a5544b5dbd0a78c92b930148 : SUCCESS in 2h 26m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ad22be085d458e8eceeb3c5c01c45b : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ce817290b85e9ff6498bfbec83fbb75216e50353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 22:06:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2cc6c3f99b1dcac727b48976006c3b4439d510d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 00:47:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafcf1f02b7247f998eeeaba5aec9938\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a6a31f160a5458d92e34ecde2d5ef01 : SUCCESS in 15m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eeb3adef934a4939954d17bb457808c7 : FAILURE in 29m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c65611dac9e4163bc0fea29d163110f : FAILURE in 26m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d8b8b36a0acf43e7b480840410a387aa : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9856a8a703cf48cb8bd8065f7ab5f5bc : SUCCESS in 1h 57m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be8ec5b2e628411699bc15e7719eb2b7 : SUCCESS in 1h 33m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/319e789d9d464716a8e5ad355f4f9f52 : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97c5e35c4ebe489587a6adcb1f28464b : SUCCESS in 1h 47m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cc1f9810f4e4519b97a7ab0023ae3a0 : SUCCESS in 2h 24m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a0ff229e0ee4e0685e1af05a68f03e2 : SUCCESS in 2h 19m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca0a1770044430baaf227505c31c7cc : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8603cd79e973288ec9156a1a540b01e0182ac720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-29 16:43:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0c4bfdbd8ffb9951458aa681391601d885c56bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 19:07:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67ca8f9ebca4585a30a28b3849fd7c0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8841b5de8a094cc1a653c142958ee517 : SUCCESS in 10m 45s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56d9ed8270ba473abc9db7d595be22ba : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8836ac2e4a194143bce1a4f7d233f9e0 : SUCCESS in 1h 06m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/71eeee5bf5ab446b80e9c397e0409c79 : SUCCESS in 1h 42m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/83c6e1e6e2784ad99015a565e5e966ad : NODE_FAILURE Node request 300-0021098187 failed in 0s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f49852a75fab4b12ab57d873e2271590 : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55b6293c1cce496e8b4e1493509c04e3 : SUCCESS in 1h 41m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97770cb222014273ba570ff646ecb233 : SUCCESS in 1h 45m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f24f4e1014944a1c85d27ce85713183b : SUCCESS in 2h 15m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893b8ddd548d4592a4d1d5e3f534865a : SUCCESS in 2h 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e9a1936d4c4374bf533960280ece99 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ee8816ac300a2bfe06eae30efee5605261ca8bf6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-03 11:00:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"40f600f7a5a5fcb8013d9bba4af65318240b34ef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-03 12:22:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0e5256f38db943b2244181c61dfb467ef96a6be2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:24:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"478a5775fae5da28c81cdd9408550106e24dc32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:17:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed971671e2c43cab13e80ee30c6ba6d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca738418206943e7ab7f25d7684c224b : SUCCESS in 11m 51s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/47427c240c8043cbac3cf1e3c51badc3 : SUCCESS in 1h 43m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/132dce9f17b34e219fcee5b68cda8784 : SUCCESS in 1h 45m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7da9de8b6f9f44c2822bf4a63a552456 : SUCCESS in 1h 46m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa67c33311fc45ef9c3e89dc8d804374 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"727de6109c8b0f3573ca6e4c5f582d0c6b7ddce9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:17:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b6333dae1b07518774c2ec29a57893b710af29b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:19:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2d926493d9f4a1ead73fa5e2e72d76d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86ff13969cd4bd58ab3b6bcd2c1fc48 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0f9a042d594c589737bf1be1135db5648259ac3e","revisions":{"16697fc71bc47b27d396407ee9627e694945abcc":{"kind":"REWORK","_number":1,"created":"2023-04-14 22:05:43.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/54/880554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/54/880554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/1"}}},"commit":{"parents":[{"commit":"e89e2d4662b64bfc42ed9683a876fac7fda63daa","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e89e2d4662b64bfc42ed9683a876fac7fda63daa"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-14 22:04:45.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-14 22:05:41.000000000","tz":120},"subject":"Add TLS support to senlin backends","message":"Add TLS support to senlin backends\n\nBy overriding the variable `senlin_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the senlin backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I7e0910f163560ce7b06c51264edc05ba43131522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/16697fc71bc47b27d396407ee9627e694945abcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/16697fc71bc47b27d396407ee9627e694945abcc"}]},"branch":"refs/heads/master"},"0a286774664b437729f06117f01c2846e986313f":{"kind":"REWORK","_number":2,"created":"2023-04-24 22:06:20.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/54/880554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/54/880554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/2"}}},"commit":{"parents":[{"commit":"e89e2d4662b64bfc42ed9683a876fac7fda63daa","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e89e2d4662b64bfc42ed9683a876fac7fda63daa"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-14 22:04:45.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-24 22:06:18.000000000","tz":120},"subject":"Add TLS support to senlin backends","message":"Add TLS support to senlin backends\n\nBy overriding the variable `senlin_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the senlin backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I7e0910f163560ce7b06c51264edc05ba43131522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0a286774664b437729f06117f01c2846e986313f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0a286774664b437729f06117f01c2846e986313f"}]},"branch":"refs/heads/master"},"0f9a042d594c589737bf1be1135db5648259ac3e":{"kind":"REWORK","_number":3,"created":"2023-04-29 16:43:40.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/54/880554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_senlin","ref":"refs/changes/54/880554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_senlin refs/changes/54/880554/3"}}},"commit":{"parents":[{"commit":"e89e2d4662b64bfc42ed9683a876fac7fda63daa","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/e89e2d4662b64bfc42ed9683a876fac7fda63daa"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-14 22:04:45.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-29 16:43:38.000000000","tz":120},"subject":"Add TLS support to senlin backends","message":"Add TLS support to senlin backends\n\nBy overriding the variable `senlin_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the senlin backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I7e0910f163560ce7b06c51264edc05ba43131522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0f9a042d594c589737bf1be1135db5648259ac3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_senlin/commit/0f9a042d594c589737bf1be1135db5648259ac3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
