)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~921442","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~Ib5a4af5ea79868b7dd7ee7f4329317815f3b466a","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-06 13:26:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib5a4af5ea79868b7dd7ee7f4329317815f3b466a","subject":"Fix skyline user creation","status":"MERGED","created":"2024-06-06 11:22:32.000000000","updated":"2024-06-06 16:21:30.000000000","submitted":"2024-06-06 16:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921442","meta_rev_id":"851f59052519f37ee93468dcd8ec99f6c08f6459","_number":921442,"virtual_id_number":921442,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 16:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":2,"date":"2024-06-06 13:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2024-06-06 11:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2024-06-06 13:26:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2024-06-06 11:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 11:45:44.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2024-06-06 13:01:48.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2024-06-06 13:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ecadb19ff48f85b853a6af8085168962618b1ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 11:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f43ab7c5fbdb43c62998ac396f3eedfc69ac321","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 11:23:12.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"e9a660ee3b6440e782502ddc2c2cd207a914eb2f","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2024-06-06 11:45:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e53a679c57cef85539f7efbbbc883b041b28c72f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2024-06-06 13:01:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a707a0f4b609916c3e4e99793a378f258f1d5733","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 13:26:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"848454eca918e76c5ed935f321873cc3a0255713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 13:50:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37613700fbe7429a9ecbc1e7aa5da254\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e587ef79c4174694ac520ff3eedcfe3d : SUCCESS in 10m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2fa75db430c141c1ac9f0ad55944e02f : SUCCESS in 1h 31m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/58e23ee39f284be4bb9bd4b530c83731 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bc03130c5e344c1a676634fba624876 : SUCCESS in 1h 24m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4adac06164024254b9e86940cdc3ec47 : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72a6a1595897432882eaef208ad97268 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b303f88515b4696bed939b754e01c9f : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c532a8091acb4ccfa30d9145022062d7 : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aecf748c01174c5e98eec36e35b2f0f0 : SUCCESS in 1h 15m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9aff0ae231944fa9ef6516ef55d09e9 : SUCCESS in 1h 48m 08s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e3369012b6b42359d40b97c21e40c5c : SUCCESS in 2h 14m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cd93cabfc3245bca4a216793679d2e2 : SUCCESS in 1h 36m 28s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aef9281254c44327876691fe0a848feb : SUCCESS in 2h 14m 11s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24330398f4ff4ba5a37b172685b13de8 : SUCCESS in 1h 56m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2891f29a14214bd2b43ef22f85e1f48e : SUCCESS in 2h 07m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdadb53683d4543a66ec5e8e1cb0280 : SUCCESS in 4m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efab0ebc514944758cf9f4a0120b756f : SUCCESS in 1h 59m 05s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ff7b3a1c3b64814b92e8659464d653f : SUCCESS in 2h 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2283a847d2a92163fd272769c7edd1bac374ecc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 13:54:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ec6825a136285863eb633f7780c593d2ea90497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:19:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1457aee7d2449bd8ae2c7400c39ed7b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/828f9f55929b4304b6f06b9c79d9115a : SUCCESS in 12m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/49048194ab4542d1b21e395ddfbb6423 : SUCCESS in 1h 22m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab3cc4f4fa7844f788a9828c2ba484f8 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae3dc72cc8c64f5bb7a44e5d8d36bc03 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecb054e2056d4077906b70c0801cf741 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f39c36f9336649a59ce567cb1a8f2f72 : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9934cc435b64b7490d1c5fdb77a9f65 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c05aa73e9f364da0bdc6c2c95690e9de : SUCCESS in 1h 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d82d39a4b1a4aeb9e0a95dff352e965 : SUCCESS in 3m 55s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/381f6857a5a44d8f87a9df6fabf5569b : SUCCESS in 2h 18m 55s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b09eba0291984e23ab8925a84e89d026 : SUCCESS in 2h 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"09f46f72e8e86eed8e154e13f78d44500e8434f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:19:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"851f59052519f37ee93468dcd8ec99f6c08f6459","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d71eb2bc564b349230667da1bb4fe1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d5646a43fb45448920450e906ac4d0 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fced22ecab95e4569dad91d9d3bc9512c2e5991","revisions":{"0fced22ecab95e4569dad91d9d3bc9512c2e5991":{"kind":"REWORK","_number":1,"created":"2024-06-06 11:22:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/921442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/42/921442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/42/921442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/42/921442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/42/921442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/42/921442/1"}}},"commit":{"parents":[{"commit":"3190e3f0c6516fcdbcc284b4504c3371a7ec1802","subject":"Merge \"Switch nginx with Apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/3190e3f0c6516fcdbcc284b4504c3371a7ec1802"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-06 11:22:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-06 11:22:26.000000000","tz":120},"subject":"Fix skyline user creation","message":"Fix skyline user creation\n\nWith previous change [1] when type of role name was changed from string\nto a list, variable name change was not reflected in _service_users,\nwhich leads to skipping creation and role assignment to users.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/918160\n\nChange-Id: Ib5a4af5ea79868b7dd7ee7f4329317815f3b466a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/0fced22ecab95e4569dad91d9d3bc9512c2e5991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/0fced22ecab95e4569dad91d9d3bc9512c2e5991"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
