)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~921583","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~stable%2F2024.1~I64e0cabeb75bf236b14c01e0eb91c77e2d369d7b","project":"openstack/openstack-ansible-os_skyline","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-18 12:04:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I64e0cabeb75bf236b14c01e0eb91c77e2d369d7b","subject":"Update the skyline proxy config","status":"MERGED","created":"2024-06-07 22:04:27.000000000","updated":"2024-06-18 12:05:53.000000000","submitted":"2024-06-18 12:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921583","cherry_pick_of_change":921545,"cherry_pick_of_patch_set":1,"meta_rev_id":"10b63d73631a0c16688b390941a569ece69e74a8","_number":921583,"virtual_id_number":921583,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-18 12:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-06-18 08:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-08 09:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-06-18 08:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-07 22:04:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"CC"},{"updated":"2024-06-08 00:44:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-08 09:19:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-06-18 08:46:36.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"0cadb803ee4934aa228dd139d7c7cded8e24e148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-07 22:04:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1f252cb16fd9228155bd25db7969c6ba39493bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 00:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f37945e1c02440993f8104602e88e5b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ee05bd97b964ec690d9fc55b7cd3a7a : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ae27b04d29d49d1bdc263d217a1c108 : SUCCESS in 1h 04m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cee37a63490440ffa0a2c310b1d27ec5 : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/803317f2229246648ead68520b56545d : SUCCESS in 1h 24m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff25b717003842b495b3d7231216a522 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afad1275e8394e739866e3c124922685 : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7546294717494b36b539a86b080bb4d0 : SUCCESS in 1h 32m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7ad81435c0241b6809e9902c6138b64 : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb074edc7afd4bd498575501626e6e61 : SUCCESS in 1h 24m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebb69c9a5350428dbe9a5eb1c93461b0 : SUCCESS in 1h 58m 08s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a60a361aad1346169426dde226b174d0 : SUCCESS in 2h 23m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0aec2f6e0c94ce58bb340c1011eabe0 : SUCCESS in 1h 49m 09s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24838543e40548a19ff4e5fdcff2d3eb : SUCCESS in 2h 08m 51s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d184c2c4cbb04f6e834c9003aa87e1bb : SUCCESS in 2h 05m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/149c80ce0d5e4b3abc111f05cb1a6498 : SUCCESS in 2h 24m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573fad46e0944c3abb840efc3776c880 : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cc9fa7f28c14e23b763056318702857 : SUCCESS in 2h 14m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf53533a3d484067a97117552548d7d2 : SUCCESS in 2h 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"485e200d0ca066aa84620145223b66dec9f3ab2a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-08 09:19:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79195818889a655870c0074649e5ec6aabdad7b8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-06-18 08:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f7621991f7d62e72ad9d4a5dd6b55464377c28f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 08:49:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c18b56239c54948d47fc3d92f5fed4a16cc05f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:04:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1964293b32ba411b93f424d3fa5bd3ff\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03f27303213c403c879bb7aaad51da25 : SUCCESS in 13m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7339a7bde4584471b5d3cc3ff97e998e : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e7f6fee7a27e44a9983d68168972b5ac : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f655d080412e4e2d8e079250dae9d53e : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/349b6e651d2b4da1a68fcd740a3c4aff : SUCCESS in 1h 45m 48s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95e988d2e73f4595a09050eb961ef30e : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a236e862efd8455f881fd84dffa5bd6e : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/574464f3e014415a88d2d1b6c93e75c0 : SUCCESS in 1h 24m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f4078b481741659724627daf19cafb : SUCCESS in 4m 53s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cb842c16c3548f6b6b4764bac211759 : SUCCESS in 2h 16m 40s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98f9f66286094981bf1bec60ecc1d1a3 : SUCCESS in 2h 24m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8110fd1085ee070bbeea84fd7bc407245e591360","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:04:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"10b63d73631a0c16688b390941a569ece69e74a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49926d388254e2e99b851ed1317f89d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04efa8c8ec7a4326996da794853296e3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49e70b6511812b61ebe1d61107ab5413b79ec7ba","revisions":{"49e70b6511812b61ebe1d61107ab5413b79ec7ba":{"kind":"REWORK","_number":1,"created":"2024-06-07 22:04:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/921583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/83/921583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/83/921583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/83/921583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/83/921583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/83/921583/1"}}},"commit":{"parents":[{"commit":"ce7dbf49373277e80aacc970cb20fb77d7d605e2","subject":"Fix skyline user creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/ce7dbf49373277e80aacc970cb20fb77d7d605e2"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2024-06-07 16:59:12.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-06-07 22:04:27.000000000","tz":0},"subject":"Update the skyline proxy config","message":"Update the skyline proxy config\n\nThis change updates the skyline proxy configuration so that it is more\nsimilar to the nginx base config that skyline generates for itself.\n\nChange-Id: I64e0cabeb75bf236b14c01e0eb91c77e2d369d7b\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n(cherry picked from commit 60e7b4fc8aadf6dddc021ed1077da67f4662f428)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/49e70b6511812b61ebe1d61107ab5413b79ec7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/49e70b6511812b61ebe1d61107ab5413b79ec7ba"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
