)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~929697","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~I3288c6e4060fe954b2c2725865783aa18bf6bc0d","project":"openstack/openstack-ansible-os_skyline","branch":"master","topic":"osa/apache_mpm_alignment","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-02 12:53:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I3288c6e4060fe954b2c2725865783aa18bf6bc0d","subject":"Ensure that selected Apache MPM is enforced","status":"MERGED","created":"2024-09-17 18:24:14.000000000","updated":"2024-10-02 18:14:45.000000000","submitted":"2024-10-02 18:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929697-osa/apache_mpm_alignment","meta_rev_id":"8bcc1ebd4e406c0b126c92aa658ca7337976fd93","_number":929697,"virtual_id_number":929697,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 18:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-09-25 07:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-23 10:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-09-25 07:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-09-30 11:11:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 22:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-23 10:44:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-09-25 07:14:02.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b6cb26b9237a2dfcde0b1a23746efcee6bfbce8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:24:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a458bf5534b1a9c5c053a70bc3af4ee5b4b3870","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 18:25:21.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"b0fb06c8dc8f3baaed2788d1e282a9a439210951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 19:52:13.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6a757f1a94a2607c22a5b6228b7d0e410df8c6f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 22:04:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d59a6befc55e4336a6d956fe7534cacb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de01fe10e954468aa128f4e8ceb49a01 : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/63da873ac577436b866f767ed64f5d71 : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/306cc3a2597b43ff8f6d837ca27845f8 : FAILURE in 1h 00m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5af3aacf10b144da9aa467114a3e5c29 : FAILURE in 1h 03m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2291189e09304aa0862d3b7d273c11c5 : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2af866bc204f4c6a9ffd04bc9da14ce4 : FAILURE in 1h 02m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9c93bcbd2c84f928182f3315ce9701d : FAILURE in 1h 09m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e37fddccdf344a4c84e9eee271dd8f0a : SUCCESS in 1h 15m 35s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d081a64f449d4ec6bd314fadfb476fd8 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb00dfd224d844d8acca1cec23d39fc0 : FAILURE in 1h 02m 19s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c0ad05e7a604578a5931f0ef53ff545 : FAILURE in 1h 17m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/062ae70f77a54239a975c71bd5a098d7 : SUCCESS in 1h 07m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/963d355d7bee4993ada25d9b71429af1 : SUCCESS in 1h 35m 07s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df7141e4286b4179b2906dd42f6419d6 : FAILURE in 1h 42m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18bc82f086e04c6eb470feeb8097541d : SUCCESS in 1h 52m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a1d492976046c08039361cacbb1878 : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8dea621a80145ffaf2dc83edaf7e93f : SUCCESS in 2h 08m 31s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f836ca61e90e4c0b8097ee511cad2d9e : FAILURE in 1h 51m 29s","accounts_in_message":[],"_revision_number":2},{"id":"126f4b9a1f1af086f15feaa10a7e87ece4bf7f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 08:27:09.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6bbf3b3d56f5cc556abf0e96154b7d3fdbadb175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 10:44:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90b20fa7393b474ea456a2001a85511b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac1903ded061456d9a1af9b19dbd30c9 : SUCCESS in 6m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ac3eb2a7fb84a4f93f376acbd97af87 : SUCCESS in 1h 02m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/85518bcda58b444fb28a5ffd743a7df8 : FAILURE in 43m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13d76b8a0883437fa6543fb2f9c00942 : FAILURE in 1h 10m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa08a7f676274b4ea24e0b3738c1c477 : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2be1970af19c4617a12a2f85cfa43f07 : FAILURE in 59m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1043517c92e2404fb58973206a68885d : FAILURE in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0083b8ab4c94cd5a557e217ff3fa3cb : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5a4cb4be0a04910a91e84e2ecb93688 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ca46b673bb14dd5bbe39df41e2a5f63 : FAILURE in 1h 11m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c53235a5611341ba83f8c6a7532fd880 : FAILURE in 1h 01m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4834b20be89d428abaa5ac5f23d1cf5a : SUCCESS in 1h 43m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22cd3cff2fa0459ab83a3317d740c3bb : SUCCESS in 1h 04m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e827a82a4bdb4ede89cb78e7d5026360 : FAILURE in 2h 04m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/899abe223f4c4aec9edb9d8309b43413 : SUCCESS in 1h 53m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135d64249f544c6f877c3b983c51660d : SUCCESS in 4m 53s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c18624e35d0407aa54ec525ebdefe52 : SUCCESS in 1h 58m 56s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/746cb6beebe24c53b8531b2029fe8a4b : FAILURE in 2h 08m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cb13fab01d08c599ccecaa93614fb8a3b6eebf88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 13:30:25.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bfe3df28121d3dbe039f61a281f10edf38382bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:44:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ae905026c2d45159a4488b72a3729e7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/518018a99d2840719de0242a5c529d61 : SUCCESS in 8m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a378e0358174916adfa7511acf5f76b : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ff7e1efa8a484988681f9c3b4b0e1c : SUCCESS in 49m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c425a0266f1a4a8c96ecfb87f5863b77 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e3a7f4ee2a5463683313aac7aa8a6e3 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75221b9c099b427a8eda9361cdddf485 : FAILURE in 1h 05m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b11c9a7d7f841cbbab507e7014b01c2 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1ef55085cd9480980bf08bb9ad753f5 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba4111d30a2455a952bc39f68fe290c : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54468d6ddeaf47f79cda141993a018b7 : FAILURE in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd650f6b181f4439adc40182e9c4e868 : FAILURE in 1h 10m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/871830e86bf2495e8daa5d6f4e42ec5b : SUCCESS in 2h 08m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e25d3f0711b048f1be6fd444642f7865 : SUCCESS in 1h 52m 31s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4499bb4ecf4d454ea281c5673472b18a : SUCCESS in 1h 57m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aee51ed66ec34e249bc1c0d019c779c2 : SUCCESS in 1h 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1513d55e2e75470db543d1c6e308bdb6 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db6cf9700a9b491c8f39e8f4a935ffbb : SUCCESS in 1h 58m 43s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1bf575c139d64602b2b3ce77db6d5c13 : SUCCESS in 2h 10m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ba447cb1f1f090813394ab714e964971268cb5dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-19 08:29:10.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cfb1ed5858e59e699e531005bcdc05a8fce641f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 11:00:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d459e65b8b44a1bb3840d2026a99d0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd8d304ca1e343ee929658800fdb7539 : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee660af6e0a84ef0bd46bf30e6a719b7 : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6bfd6ffac94f089b66e99667caf075 : SUCCESS in 1h 19m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8d2ade7fdf74a518f69d067c1d6e60e : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6caaeb1ed1b48aabf32056bfd597c3f : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef341741aaac45c9b1336e62634f91e0 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce85558a378d4fada70cf88caa1c5d96 : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9760c066ae7433e8903e1a9c812512b : SUCCESS in 47m 38s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ddfcb82003c437ea91040c2e2b3fc5a : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64ccb39757fd41f8adffba7e6f2df9ef : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b28b63230504848a2179d568714400c : SUCCESS in 1h 35m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1202434167aa42d7b0f782c2790c1cdc : SUCCESS in 1h 58m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/939b454462cd4a838df1fe493340a66f : SUCCESS in 1h 49m 08s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19ad48d368b640f4bce33b74c83a1256 : SUCCESS in 2h 02m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/957c25775a6245a2b2a0eb100f51e0c4 : SUCCESS in 1h 55m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e769842c0c9476fae7df7ae6c599736 : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637d3f28a5f5416eae1a44effc183e6c : SUCCESS in 1h 52m 49s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05976231966b4d4d87d6dbcdbf774af5 : SUCCESS in 2h 19m 07s","accounts_in_message":[],"_revision_number":5},{"id":"91dd6f47c68bdca5620f413d437876cfdefef25f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-23 10:44:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"278fd28e4cd8167be06bbcad7c0708fec8ddd9e2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-25 07:14:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9e98501cd9b48f3a41cab693a291f97907be41b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 07:14:14.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf61d0e8f00d421cadd94a14a9f713ae\n\n\nWarning:\n  Change 929892 in project openstack/openstack-ansible does not share a change queue with 929697 in project openstack/openstack-ansible-os_skyline","accounts_in_message":[],"_revision_number":5},{"id":"84879b3bae10d125371e4e15328e082907037ac2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-25 15:49:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21a79d6505f0a3ba04354354572b1f96308c953a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:15:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8131cdcfee25423e9713ca030990b1c0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c340d7b2991d4d42a55e019380f8bd35 : SUCCESS in 6m 37s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/baf564d126ee4dfda11f2b6acd4d1e0b : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd64fba6e3fc4a3fa10fb151bf1b34ec : SUCCESS in 1h 18m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8961f360d098426baf840511761a38b6 : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8ed5f8fc318407eb1121793c07a24a5 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74b7369a55254d049ff00f1da1f37b9a : SUCCESS in 48m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32384033b8f64aa6986fe8bb0d4ae15b : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/003889b3c0ae41f0bc73cf19081f2dab : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2275a799f38b4d569cc3e7cc452dd3f7 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffffb97470f54dc3a346ff2f3318f8c3 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11d40f6ca27e4dfbad1dcf7b14a2fbe4 : SUCCESS in 1h 20m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a799a0b553d34803b8cd5e47ec8f7c56 : SUCCESS in 1h 58m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fddef9e6f6341cc83bd4a9ce7cf2eac : SUCCESS in 1h 40m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbe34e9899ba49858f97a368c2b2fa99 : SUCCESS in 2h 07m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6268e75aecf44c19c719ae4f6a005a3 : SUCCESS in 1h 52m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7764447bbe9d4247bfa06a60014f7ae8 : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5329df05b2b45608a2763a2f32a4474 : SUCCESS in 2h 03m 40s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/760d01457eae4dabbaa92645f57f013e : SUCCESS in 2h 12m 04s","accounts_in_message":[],"_revision_number":5},{"id":"953b1cb3c8c333b61b2f118927fc3cd07c8bd1b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:15:30.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c09f5a08815048fbb8c792aa5b9f9059\n\n\nWarning:\n  Change 929892 in project openstack/openstack-ansible does not share a change queue with 929697 in project openstack/openstack-ansible-os_skyline","accounts_in_message":[],"_revision_number":5},{"id":"ebb7251bdccfdb5923581143e4a610265f46c753","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-30 11:11:08.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4e98f539d499de6770e679cc9dd3bb879a6580e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 13:53:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9743c0c2b78c45fe9143c431ef0a7c74\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36db81b11ff549ffb412ef38e58c3911 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bad1c00b44464c66baebcea622405245 : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/81e0809c89614bf8944bf1178a208e33 : SUCCESS in 1h 18m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/61fb7b4836f345068a54f457837eec15 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba267f45691b47278e2bdad493a934e2 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8728fb91911344fd942e23d8c448db28 : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c82165fad5948ebad24e799c68d7bc6 : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e9f3915924b44cba7dbe4af154b6375 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/423d7df3955446708818672342e27a00 : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13c21c48e6044937a76370d201c3165a : FAILURE in 1h 14m 12s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb77992d16574dd588e8aedb8abf4463 : SUCCESS in 1h 26m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba851598bc15490ba038e4946f32b813 : SUCCESS in 1h 44m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa389a18e13b48be85c460a991ea883c : SUCCESS in 59m 08s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2781f63e9214194bc8457ce62b70240 : SUCCESS in 2h 39m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9db325f6b7b4d51985a77ce83f16952 : SUCCESS in 1h 55m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051dc1ca3ab9481e88b423cbf410c55a : SUCCESS in 4m 16s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/013a8bb488be4b6da8c17b5130c14966 : SUCCESS in 2h 12m 36s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8a1055256004be6873961ddf31ecb8a : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":5},{"id":"9df79c7ccc9f127c2c675a71c2dce626edc8a122","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-30 13:53:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1e2ceae2a2bf52cc83aad0ce31ee3ba9b231946e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 17:43:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb0114709b234e6c8ca5e3d6fc46cace\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/661eea47e7d64c79a1266db378059083 : SUCCESS in 10m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6822b0e78bd3488ca9f8017c9f09e0c5 : SUCCESS in 46m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/04e2925c8b314223a70196add23c28c4 : SUCCESS in 1h 13m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf28c28c82164d93ace8a72bb6c865b3 : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f314c0f772d4f62ae0b04b4af02a987 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2016acb2a81a496eb0688c26eda02954 : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f58b685ed24499296ffe844631a689b : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671787436dfd47599fe4f993a888e516 : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6136ee2793ba464a8ca24c107902bb07 : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77e7089ad4304c64a49cbacf809e44b2 : SUCCESS in 47m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba450f139b0f4d5dac1434c66a833d4e : SUCCESS in 1h 23m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11a6a65f021b4fc7b3b41b18a24ba386 : TIMED_OUT in 3h 04m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/255fd9fea69f4c2eb92b61f90eb3bfe2 : SUCCESS in 1h 51m 12s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1724bf03ac8a4ae0951523ca34f1cc7f : SUCCESS in 2h 04m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d403958f7974468a216e23743dfbd3a : SUCCESS in 2h 00m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3338f195e53e440d814c3bfeb99bddc5 : SUCCESS in 5m 10s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de7355f727d849b3b9a387200594f3c1 : FAILURE in 1h 40m 26s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/207ea769183e4ba288df660257478336 : SUCCESS in 2h 18m 11s","accounts_in_message":[],"_revision_number":5},{"id":"5ed7d112fc8b311001ede9ef91b7c2edef60a671","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-30 18:44:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ec54eeb8beacd7c59dd95057a1b51399d8f49d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 21:12:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428b57f46a374182998cf8834c70999b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce0310c7d4e7412489a252076a6fb982 : SUCCESS in 6m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ff1fcb2b5eab49a9a5b00fd4fee31090 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/320a846aaf304e1e9f3722f50034d03f : SUCCESS in 1h 27m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a655506f49a4a0b8e520727406e8f30 : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc85b0fdff4b43b0ba2758c692ba339f : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ffea3aabe6b4d49a937ab0139320a4b : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4819943251f4305ac6fde85da1462f3 : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/376e6b05380f4015bc6b4406c1f0a334 : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1a47fdffbe04c24af2b9e5895e24e2f : SUCCESS in 45m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e82ffba3e95447169d299b2ce103abd6 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd3ad7204c7543a99b2906e2412d7725 : SUCCESS in 1h 14m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d31c4a05748495480795b9315d688f2 : SUCCESS in 1h 53m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5d9c25d7b134dfd84d62245ba7081c8 : SUCCESS in 58m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27a7f929207546c8b6cfda95ad82a127 : SUCCESS in 1h 50m 10s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c2f06295ae849e8a834903456691c69 : SUCCESS in 1h 41m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cfd5164d7440df9a64ec39e70b6051 : SUCCESS in 4m 24s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8256c19c34c44526a0cbc7a0f8cb954c : SUCCESS in 2h 02m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6317bdfe90d94072b3377ff6728bf408 : SUCCESS in 2h 18m 44s","accounts_in_message":[],"_revision_number":5},{"id":"01fe0800850eb3e0de40f359a3f28d04e62fa781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 21:14:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1b1c82f85ed9d74ffc87be92a2ef8ae83bcd7eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 23:27:31.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52ca526c3d2c464c80d0fa1eea35565e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88afc00f7c7b426a85494d288efd0689 : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/eb472b3c38014f5d97d564cd2930395a : SUCCESS in 1h 26m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/beaed8593de94eac8eeda2bd7ef1c870 : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/948c16ac1fd44fa3b9388fa9200bbd26 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c59e7f2bb54e9894e02852854f1144 : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a71689c78ad4d1b84ff27fc34898d86 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c9e871bd2f241a79d973ccb2f4bf25f : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ea77ac052534a3b9c6f35a18a286283 : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55a74a392bee4829aff57b2992cc96c2 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a2597fa3c714fe7b423b372d50fe15f : SUCCESS in 1h 23m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25a18de1e0b4e68b9fb50353e37fc77 : SUCCESS in 3m 35s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cd73f0f12d844568159e17f6be744e5 : SUCCESS in 2h 07m 51s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08e8808f2989421e9088059fdb4a65a0 : FAILURE in 1h 36m 26s","accounts_in_message":[],"_revision_number":5},{"id":"7f4c4741ae93bd8a7d9ff52a87fec3373e3a587a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-01 06:56:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f1008f970d0d298f72a8777c20d63270ca22e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 09:11:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12bda3b13db04a73bfba811bbc80a43e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/252346438fa84f90aec182bae02a7d9a : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/38892ad16b05495798ba300e04f1584d : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c097e2e0509d4e4684bbf871233474b4 : SUCCESS in 48m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea9ed5ee887b4ac2b7ac5da1bfbefd55 : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6c354e1461e4ea9b501ffda3bd70427 : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a90ee712f80b4167b41de4b624a6eb7f : SUCCESS in 51m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/987e3d3fd6e741b2935c9563f1c39300 : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00879ec873a145f9819e00933e8c0242 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eba5747626324b34a94f0039b84ce936 : SUCCESS in 1h 15m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af49298f0cf1462ea30b1e4f1ebdeb33 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28aad5b17e85477dbacdb3e691cc2bcf : SUCCESS in 1h 10m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10b18d647b644b76b69ab45d0edec875 : FAILURE in 17m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b4912004bf7471e858804f2d895e936 : FAILURE in 15m 55s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ccc0c00f3e84af69977d7933f8e2e52 : FAILURE in 17m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74e6b98fe020495f8736e029ef987ec1 : FAILURE in 1h 02m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c536e53ebb048709fd399712da7b331 : SUCCESS in 3m 53s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec3aa8c41bb84c01b79ec603d3a924ab : SUCCESS in 2h 07m 35s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/60a6ff0746134984833a2cd9006c21b6 : SUCCESS in 2h 03m 46s","accounts_in_message":[],"_revision_number":5},{"id":"53712f3787942301014221d5d80997d898507ba2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-01 15:16:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2555495fc97b8d3183bc7a051d8a4bf5ac79c6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:31:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f74f8903b89463d958deef8699df1a1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e567927911bd4c58843e9582d7b084b4 : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/07cd224c7f4a46caaad260f328ae725a : SUCCESS in 1h 23m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/da423f878c9c4f608cdb6b225f679608 : SUCCESS in 1h 24m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3afa07a3109f4b2ba5f1eb29b962cfe2 : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/044afc73cb1d4a12b832893eb8f60497 : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57c7a51f26a24ca385feda6228fadc95 : SUCCESS in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0dca57dacb3a4bbfb6fa0a352e20d490 : FAILURE in 1h 14m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceb7175f9e3f428d94a4c99805ba8f67 : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9299ea9258d74fc0aac56946dc3379f5 : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c667e517177c4c7bbbbac21a394b21eb : SUCCESS in 1h 44m 56s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e4d2764e4cf493d80ab36ccc800df73 : SUCCESS in 1h 28m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1742e0a7248c4403800fa70feb1add2b : SUCCESS in 1h 49m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3591ffba32b34a9b8864aea27fb1b10c : SUCCESS in 1h 38m 25s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/069acf287ef746aeb8e7aa573dae8d7a : SUCCESS in 2h 07m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8089bba82a64c82be92bd1f13929bf4 : SUCCESS in 1h 39m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a105ff7194fb4d3ba57c342f937ece61 : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3ce5cb4158c40258c27b660a722d7c0 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55bcf85f48fa45a488505ffeef5a54b8 : SUCCESS in 2h 07m 02s","accounts_in_message":[],"_revision_number":5},{"id":"b4cc8bcc1e63cbf2e1a41d92c8689a9c589ab989","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 09:53:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1b68e422299b1fd43a11675ded88598fc7b87745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 12:44:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08e8e13719b34ddf99cc67d282f966a8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b30ef9b077c4ad78a8ebf0880b09202 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed6ceb9f4fda4d0ba419b16292694c3f : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a523a9310afc4e249e72904bb983b7df : SUCCESS in 1h 18m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e56401bc48394407831595a47443c9f6 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/125f4ba694d74270aa9bd1e002805429 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faaa35cc9b1d4266ba9b82ca9565fdc1 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad5f3ece61e84c978ff1bc89e3b0c3ed : SUCCESS in 1h 27m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b24755871ff24a1ca94165bb5f661c79 : FAILURE in 58m 43s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43b8188d89be49cb891552ca8824df8f : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f84cfd7c9d94a41a64bfd0f6a90ea38 : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fededbb1b48142b9953ec30db72b2fe2 : SUCCESS in 1h 28m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0cfd58029c744ba99bb6b47f6f9d2302 : SUCCESS in 2h 02m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46f4172184934076b293bbe56512cefb : SUCCESS in 1h 55m 52s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fecbab6c680b4920982f8ac4c9873a04 : SUCCESS in 1h 53m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92e173f9f63b4da7b8869a940a8f304c : SUCCESS in 1h 44m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d7a496f3314eac93550fa0b15e9f3f : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ddbea2c6521457ab04afcc5b007dee9 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5fea9b790614453c956ddc423b8cf057 : SUCCESS in 2h 27m 46s","accounts_in_message":[],"_revision_number":5},{"id":"63f57289e3ba6574ec6fb72e87167671471f8369","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-02 12:53:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"335d9a8372bbc2d2bb8b7b08be67ff738f696f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:31:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e358a33bc841e39b6651481cd5162f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9222f948b1624222bd4f7fde02e8cbf9 : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/beadfffa1871479aa63e58e38a75d792 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4baaf1427dfa4b6bbe5b9afc3f8aa4f4 : SUCCESS in 1h 18m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d33d40bff694419909ec525da899262 : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b45bc4c2d69f460b9af8807cd7c7e1d2 : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b072004fca54beb87542df420d513eb : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb01b2bcfb9b4ea19e4f43e87ed1378a : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/976a8dd86b3d4236a325f918dfa409ba : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b9e3a69153c439ca76c4c0669f3d0cb : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/455c4e8b5f744204b58f0249b916def8 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc2df81d5a0d44e99419eae47915b526 : SUCCESS in 1h 31m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/620abb77997345caa975f2222d588e34 : SUCCESS in 2h 02m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f7e19ee24145788d4c3aa06e35082a : SUCCESS in 1h 53m 47s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b1bd6b2d9044156b6cf8d4057dcc9bc : SUCCESS in 2h 04m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b49b8a8c5fb34da7a29430f239740f15 : SUCCESS in 2h 07m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9e74c67ed6430e92230bc1af83f014 : SUCCESS in 5m 06s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f97f3d72d1004e6fabc9291d336b4cb0 : SUCCESS in 2h 15m 41s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/77628637059d4bea981c95dd83ebc91e : SUCCESS in 2h 29m 14s","accounts_in_message":[],"_revision_number":5},{"id":"85887849899afd9d14b3a5069f76f1347ac1c871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:33:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fad2215274da87cf50261f78b0b4f64c4d4cba18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:13:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/babd3eb000f74875b9f5db37bc0b653f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227880f5a08147018befd5dd42cbe47e : SUCCESS in 7m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/13e25ed861be4a80a88cd2ba25af7a82 : SUCCESS in 1h 29m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a583a09f03124d26887a88715ae76bed : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7e28e1bdbe84320bb68ff84ecc2b22b : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26ad6e23373d408d82342402fcf19dd7 : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be395be0111049c0a1addee310daf90a : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/597b1c08a18a4e909dada16916080141 : SUCCESS in 1h 10m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/828969f761fe4591a732b6f58bad9da6 : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/534d82e0ae7842fdae5de91e16e3b897 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f496df9211764c51bbca36e8de26639f : SUCCESS in 1h 23m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeafb4ea3d424246866e12eeec0a9e56 : SUCCESS in 4m 23s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/627c0b04bb9343358f60584a3dfd04a3 : SUCCESS in 2h 12m 09s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9e29005e2d74ce9842a813ba5d95722 : SUCCESS in 2h 21m 17s","accounts_in_message":[],"_revision_number":5},{"id":"49016e796315db88385fd04cb54593de05797a81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:13:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8bcc1ebd4e406c0b126c92aa658ca7337976fd93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:14:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42104ec326a34972bf7e6bdce357ec7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2984c1c7fa2428cbf0ea6a0d7f327c6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7b19e16cbb076cb0d85c4a592750580a9aadf596","revisions":{"cb3521b57561476b106656f91cab065282ffe717":{"kind":"REWORK","_number":1,"created":"2024-09-17 18:24:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/929697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/97/929697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/1"}}},"commit":{"parents":[{"commit":"93b7ef9b2546d0998880877c682b8d2dffd43e4c","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/93b7ef9b2546d0998880877c682b8d2dffd43e4c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/cb3521b57561476b106656f91cab065282ffe717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/cb3521b57561476b106656f91cab065282ffe717"}]},"branch":"refs/heads/master"},"4529d077aa7cba5525d275a9158026fd84becebd":{"kind":"REWORK","_number":2,"created":"2024-09-17 19:52:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/929697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/97/929697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/2"}}},"commit":{"parents":[{"commit":"93b7ef9b2546d0998880877c682b8d2dffd43e4c","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/93b7ef9b2546d0998880877c682b8d2dffd43e4c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 19:52:13.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/4529d077aa7cba5525d275a9158026fd84becebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/4529d077aa7cba5525d275a9158026fd84becebd"}]},"branch":"refs/heads/master"},"eee13e6d7885509ae98b5bb3bd41c0a8dfc76615":{"kind":"REWORK","_number":3,"created":"2024-09-18 08:27:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/929697/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/97/929697/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/3"}}},"commit":{"parents":[{"commit":"93b7ef9b2546d0998880877c682b8d2dffd43e4c","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/93b7ef9b2546d0998880877c682b8d2dffd43e4c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 08:27:09.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/eee13e6d7885509ae98b5bb3bd41c0a8dfc76615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/eee13e6d7885509ae98b5bb3bd41c0a8dfc76615"}]},"branch":"refs/heads/master"},"d91f4d37b09728c33e29162b95aab7d8b49612b5":{"kind":"REWORK","_number":4,"created":"2024-09-18 13:30:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/929697/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/97/929697/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/4"}}},"commit":{"parents":[{"commit":"93b7ef9b2546d0998880877c682b8d2dffd43e4c","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/93b7ef9b2546d0998880877c682b8d2dffd43e4c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 13:30:25.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d91f4d37b09728c33e29162b95aab7d8b49612b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d91f4d37b09728c33e29162b95aab7d8b49612b5"}]},"branch":"refs/heads/master"},"7b19e16cbb076cb0d85c4a592750580a9aadf596":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-09-19 08:29:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/929697/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/97/929697/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/97/929697/5"}}},"commit":{"parents":[{"commit":"93b7ef9b2546d0998880877c682b8d2dffd43e4c","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/93b7ef9b2546d0998880877c682b8d2dffd43e4c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-19 08:29:10.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/929892\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/7b19e16cbb076cb0d85c4a592750580a9aadf596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/7b19e16cbb076cb0d85c4a592750580a9aadf596"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
