)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~931256","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~stable%2F2024.1~I3288c6e4060fe954b2c2725865783aa18bf6bc0d","project":"openstack/openstack-ansible-os_skyline","branch":"stable/2024.1","topic":"osa/apache_mpm_alignment-stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-07 06:41:12.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I3288c6e4060fe954b2c2725865783aa18bf6bc0d","subject":"Ensure that selected Apache MPM is enforced","status":"MERGED","created":"2024-10-03 09:40:24.000000000","updated":"2024-10-07 11:40:29.000000000","submitted":"2024-10-07 11:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931256-osa/apache_mpm_alignment-stable/2024.1","cherry_pick_of_change":929697,"cherry_pick_of_patch_set":5,"meta_rev_id":"63172fee3ad03332473d88cddb1dd3b47d6ad2a5","_number":931256,"virtual_id_number":931256,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 11:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-10-04 09:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-10-03 11:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-10-04 09:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 12:31:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 09:20:02.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"32df79fea07c4be0ace084847edc9c829803da60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 09:40:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a245364a3a8f457f55fc374ae8b83bd4af2ee999","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 11:07:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb745976a0a117f8bfefbc3af46dffb0a5d67e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e032e1717a0e4cd98cf8929b574ecec0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07a6dff683b345b480838c9421589f8d : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/51f209d1d24c468190edfb9513179ebe : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/088eac86e588440782a3a5030ab27c6b : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/384dc64af52b475ab5afc1db1607cb70 : SUCCESS in 1h 34m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ab32f32d23e4480580db6155f84fb856 : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a30d10c7c3f44410a6422e60ab5f6bae : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/008bd6a0b7174ae2b8ccc283feea52be : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff240a4b76d942fa883719387dd019c5 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56c197d206ee4e6e9736266df2405274 : SUCCESS in 1h 09m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7bf18f6ea14b4411898f5c6a7c4abc83 : SUCCESS in 2h 14m 18s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/015e4d12887d446489c914411721b051 : SUCCESS in 2h 29m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6726c4a531e34a319744ab6a72226708 : SUCCESS in 1h 40m 58s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5eab6196b464732b5b10a374d58f721 : SUCCESS in 2h 04m 05s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3af3d602f1ea4e1f8eb58d19e2d0cd83 : SUCCESS in 1h 15m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9439a1a98a07437fa448a890a1dd04d1 : SUCCESS in 2h 25m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60b734e78704c6d9ac18fb79e78b3bb : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acb743e8523e41b18de86af844139e9e : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41267676a88047c6963614f8abeea470 : SUCCESS in 2h 36m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3979bc6aeaa1f5c629b15d396978d76221c4d22f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-04 09:20:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bd2129835efa136f3036647a0e307d0f2134e1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 09:22:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10011d549c9caba905107f264071f97011e93760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:52:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f611a8d583c949409ac431962f798b0c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78c91e78a6184bd69b9b8c33f893423f : SUCCESS in 6m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/493e39f4e21443fb8732497bd9940dcd : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/37f5863b8187485e901bd002be0af426 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79b6a46d69524e779146b1a7fc7ea4f6 : SUCCESS in 1h 17m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3484c90805c84d86962f3997970853db : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/725b2b48fe904f43b4568744da09faf0 : FAILURE in 1h 47m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bac0e3997dab462396068f61b8929f2e : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cec00867a2b411fb1a4ff178aebcbe4 : SUCCESS in 1h 21m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87de896ac5274543955c23019043a5af : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/508e484a0b4c40e1af1a7780b219c1b1 : SUCCESS in 2h 20m 28s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4978716a306b47ada31fa3fca736a85d : SUCCESS in 2h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6f2220c28ebd624d1f17ea13a1e38feff8a1e350","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-04 13:42:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30eef1190a0dbc82b18d18355c36dd572761a054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:22:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/477a8bee18d8417fbdb52c7bb6013981\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbae464913af4d7091931480de05e59a : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0b64b00ae4784f448278da53e609c656 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e0ef0744451c46e18ffdc7a9d66c7ee9 : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bdcada71bf34d2ca9b1f300fb8832ea : SUCCESS in 1h 23m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2eb2f76cf112482e91734dbcf697980e : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eb4a7b04a554edaad748deab04a32a5 : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0aa8ca253f714498b1c110b732f2b971 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/427930f96307486d9766204ff5aafa6a : FAILURE in 31m 45s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9033eff8cc5b4dbaa75dcf0be6c619b1 : SUCCESS in 1h 16m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7db3ad5409ae48f5980bb9c50b2f4e26 : SUCCESS in 1h 47m 59s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7e7733cff3a49b5872a3c913108a0f9 : SUCCESS in 1h 23m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/772c147c7ccf4e51a91392e73b52f7bc : SUCCESS in 1h 53m 00s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c21522f8f07e44d19a8689e62b402b29 : SUCCESS in 2h 31m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f29eefe29b1341fab0420ecb30337e80 : SUCCESS in 1h 14m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c213374a77249a98d279ca63c3a804b : SUCCESS in 1h 56m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294a667e07404f238eb9b80aede1614d : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbdaf00354c34770b3d691ea239240af : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/918f00b13ff84d0aa0f72ba3a768374c : FAILURE in 2h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"990263da145b379efd2db4dc47420d016177629e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-04 21:55:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdaa638d13dee95079158e57bd8fe978e6dd79d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 00:27:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aced8b58a7a84067b8de920cb0e3dfa4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/786071c4b07a45e38b120d7e06aa56ea : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0d3532abc8344db49df9802e47dfd76e : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9e91c180365d48c4988e1810777a3e68 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d478f2f6766c46ffbfe17d2562fa0f1c : SUCCESS in 1h 20m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d0d2e41adc54ee9b369730a141e17c5 : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c49fcc9f3d8a45d3ade5f90dbbf13661 : SUCCESS in 46m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db7c04951ad84888bbcdf72d7982d127 : SUCCESS in 1h 24m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/303802cac6c44d2087830134c100f9d7 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f59259255f5c491b9fe77b86dfba9e72 : SUCCESS in 1h 04m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8ca7f04abc749a78773ceec332113df : SUCCESS in 1h 47m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c773af21079c471cb6326f2b05a89147 : SUCCESS in 2h 18m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6b72ec8b1bc4f1eb590c66cb462b439 : SUCCESS in 2h 23m 11s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeaa74969ecd417688c9354e85eaa93a : FAILURE in 1h 45m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3d9c3ee4c2c463e80374955fb6222d3 : SUCCESS in 2h 07m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca0d86e83f75475d9775f1ee5c991f8c : SUCCESS in 1h 45m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e0be58286f4108854a6ad4fd611baa : SUCCESS in 3m 57s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6948c2148fb4655a76613c729432ce6 : SUCCESS in 2h 04m 01s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b555b709a624e4fb72d3381ae8b2cab : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"164c1deba51a9cf744ba1c762d443c663e04ae39","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-05 07:51:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4852600a451c5bbb64793e605687a1e384bfc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 11:07:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39c83ae2ee904b769eb15ad0b8a35078\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/953aa425b45442979b436caa5193e923 : SUCCESS in 6m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9b18e48365d84743bf7cca8ea3c02000 : SUCCESS in 1h 21m 37s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5d0fa4e78e144396805fd0436fa1e48e : SUCCESS in 50m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7436c331637348ed88fc24fcf185ed3c : SUCCESS in 1h 18m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b8db64dec524298bf967d85e990f690 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a729e9d4ff544f4bab610c491420ceaf : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df78cab763de466a9472892ab5d61adf : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfc5ce65c30943b4b5c6ca58d3280608 : TIMED_OUT in 3h 03m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a19d5160f8804c129dcd09b586671b79 : SUCCESS in 1h 14m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68c1317629474e3780f604ec417a42e5 : SUCCESS in 1h 10m 30s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c6a12e8f74441e69470d5cebe07d3b8 : SUCCESS in 2h 01m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dc43f792d694fc3a12d9731626ad30f : SUCCESS in 1h 56m 59s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/710bbde4eb6f42c7a9d126f26adb823e : FAILURE in 2h 23m 48s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4490e638ef824b3abbf56d3d29d60291 : SUCCESS in 2h 16m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b2fa016add94081b242c7a69b0fa57e : SUCCESS in 1h 52m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aef3bd1d44443669ce47d6709da14f2 : SUCCESS in 5m 40s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9503f188c9d74bf79615c679cff8a503 : SUCCESS in 2h 12m 39s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3fda3dd009d647289d326000ef70396a : SUCCESS in 2h 35m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d742238b39e7618cf9937c1d3d8cc4c7d760f24c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-05 14:44:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ef9d2a144d2dbea5b773c6d45dd51de83f82499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 17:24:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/952b4a92ec724be4bed0560f2c2ed763\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/247dbd4127a14b15a51c6aa880adefda : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/187773f4fc644f80991348503f207a12 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a5d250fdc554fb5bb068aa62689ead2 : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7da6231bff48f6832b6ffa126ae560 : SUCCESS in 52m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/159656af6d874e04971910ff68678b39 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1593cbebb7db4263bcff82eb85196015 : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc45635e139e469abbeef90aacbbbbc4 : SUCCESS in 56m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac30c8133a046a4abeb6cd3f6023418 : SUCCESS in 1h 48m 46s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/693bc2fe71be4ecba30bf3bc7ec61d6a : SUCCESS in 1h 32m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bbc1244d0fc344c0aec2bbb5b8ae45d6 : SUCCESS in 1h 15m 45s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ce03aaeb3b04e23bd15ccb5342167af : SUCCESS in 2h 19m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf4c68f7428341fc8ef9323c29d2be91 : SUCCESS in 1h 48m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d93cf24e3d3b400d964aceb6d476a430 : FAILURE in 2h 25m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cc50570370a4cd882498a69d2c0fe12 : SUCCESS in 1h 19m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de9e19ab83924d409379473ba84efb33 : SUCCESS in 1h 46m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a837f0a6339842ad9b09aa30baaddac4 : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c035a6890b484f708d2c8404a09273a1 : SUCCESS in 2h 27m 04s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a6edbff16de9472a9aa8b5232f9dfce5 : SUCCESS in 2h 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2f88409814e537f73a83c1952b79587f5dfde3ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-06 09:12:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ca0478bd74db2720f329c5c1dc39948aacf08c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:45:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/038ebe298abd4e38bbbe145dc09796c8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/befebf65b7fd4995b58e96a830dc1ae7 : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6f87792528364844abb99e66b1e027ca : SUCCESS in 59m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e65e4c348a4b4aa3904b998ecc0236fd : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef3e39981e324936978b867b20bc1979 : SUCCESS in 1h 08m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f9789ff1ea34471a8f526821bc3fd9e : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24a47ee21563498caa5f0ee9e0b06518 : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d648500716c943409a72cb87be12ae5c : SUCCESS in 1h 10m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e98273cbe9c42ea8291adb44af2b44c : SUCCESS in 47m 32s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d23810e2537432490cf7a2e6b416a9d : SUCCESS in 1h 05m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ab1fde00caa4d978d2b9666e8a61dff : SUCCESS in 1h 38m 21s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10cda09346994bd1a7a5af5b0f5931cd : SUCCESS in 1h 55m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b268e5b6a3424121a3d43770f51d0a25 : FAILURE in 13m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0747fcd9a2f54dd49e3367681d1e7210 : SUCCESS in 2h 30m 54s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3577a0b5523c421e92fc0cfdb17acc81 : SUCCESS in 1h 45m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9204263b2ae74bd088b52c48f1c2a691 : SUCCESS in 1h 33m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd37db5dda0444db1d455ca0ea117c7 : SUCCESS in 3m 56s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29ebf89d8c204f66bfc603aa6a4b5a20 : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47138093b47947b78cd19885f3bce7c4 : FAILURE in 1h 09m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a2660050ed333854790d0d95d3e641f5b47e6847","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-07 06:41:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"495c686a7499ee14fc504a0e6fdf2bff47b1fe10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 09:17:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55bf008cb0be4cdf87d3c2d780e4362d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5b14050285b4dfa9d043aa6be0864db : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc89860c2196479bbd4b1fa40bb3d0a2 : SUCCESS in 1h 20m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9bac99e2427746de8d977c1e855e7512 : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aa61cf016164829a865047cb4500367 : SUCCESS in 1h 23m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a16db7d6d34d4d3bb53a617eace1cea5 : SUCCESS in 1h 31m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1310523ac4e04f73a76ca6ae3636de17 : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/337491d760b346f394626b0768638b07 : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cb2f95036dc44149f36d82a946a82a5 : SUCCESS in 50m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/428e39a5861941a39a8b6938b54ac9ff : SUCCESS in 1h 35m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c79861d45daa4b85b09f3c5b83ee856e : SUCCESS in 2h 00m 12s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/663c2faf5d27401095b96a44e9f255dc : SUCCESS in 2h 13m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08d223d190764545950483c30793b507 : SUCCESS in 1h 52m 15s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fe89d5a8c2048e6b980b88d381e850f : SUCCESS in 2h 22m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc661ad7adba4a04bf730f2c41372e62 : SUCCESS in 2h 01m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d15d89272da649709c2de6a22ccf03f9 : SUCCESS in 1h 51m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80bcee64d0c46bdb5d2918b82654b11 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cc93892fbe84a249f9c0d94bbaf23ce : SUCCESS in 1h 52m 12s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/460212f84c084ccb89dcf44e63a86d97 : SUCCESS in 2h 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"857aa0cfe73eadebb8ca2ea12b7a0fdc4e77c89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 09:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d90d2abf7d95c8b27e648e543cd569ab6d4d24c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 11:39:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869fe4598d47420ab6213798ed0be649\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/032adef6f7a94b81a005d76acaa87ca2 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9df9b09a681d4257824f48d4b8338420 : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/436ef80afac04ecb9c7d32c888b626c5 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/92d8d7096dee4d0797a6981b3ecae49a : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e63ad273d37341cc8800f09b7e63c874 : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77c664a4b764fd7bd4760002575c889 : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aef253fbdcee492ea4706af50d79060a : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd41b9f55b56456e8859d52b040c59d3 : SUCCESS in 1h 20m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1b3e63a47940f69580fed8a615c075 : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/212c2ffda2b34b97acbfef4acc77039d : SUCCESS in 2h 12m 31s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f33671ea5cb440739b69b2a9ab7f840a : SUCCESS in 2h 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5afc6c62b59ccdddd3d58eea7128f2fe73d73d6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 11:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"63172fee3ad03332473d88cddb1dd3b47d6ad2a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 11:40:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ceb0243cfaa45429057183107a4b88b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7f889fd69b4754ac86f5b4e52447db : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57c43ac40ced05b7561a25cc8993718db9b5c7a7","revisions":{"57c43ac40ced05b7561a25cc8993718db9b5c7a7":{"kind":"REWORK","_number":1,"created":"2024-10-03 09:40:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/931256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/56/931256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/56/931256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/56/931256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/56/931256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/56/931256/1"}}},"commit":{"parents":[{"commit":"f8f56ddb695864dbf98a0936ad3f4b303ca95b00","subject":"Enable SSLProxyProtocol for internal backends behind TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f8f56ddb695864dbf98a0936ad3f4b303ca95b00"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 18:24:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-03 09:40:24.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/929892\nChange-Id: I3288c6e4060fe954b2c2725865783aa18bf6bc0d\n(cherry picked from commit 7b19e16cbb076cb0d85c4a592750580a9aadf596)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/57c43ac40ced05b7561a25cc8993718db9b5c7a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/57c43ac40ced05b7561a25cc8993718db9b5c7a7"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
