)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~954595","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~Ida475087736984a2f9237d3c7b1a509166d3a623","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-09-05 16:19:49.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-09-05 16:19:49.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-05 16:19:49.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-09-05 14:40:36.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ida475087736984a2f9237d3c7b1a509166d3a623","subject":"Disable Peer verification for the proxy","status":"MERGED","created":"2025-07-10 10:50:57.000000000","updated":"2025-09-05 16:20:54.000000000","submitted":"2025-09-05 16:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954595","meta_rev_id":"8e096c0ea2d1ae83f4fba37601c138cc4f756b41","_number":954595,"virtual_id_number":954595,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 16:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 14:00:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-09-05 14:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-09-05 14:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-09-04 11:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 14:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 16:17:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-09-04 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-04 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-09-04 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-09-04 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"90f97f0019da7608eb1f4b5c51773b53823f1e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-10 10:50:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6b7d6d5134535f2b0fd874ac395da2e090cb6e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-10 11:08:23.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"f5e20075ac65c9ea10679bf7b38b6b253a59084a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:31:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51015baeefec4c91a628d56141586e82\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48a50e3319be42c6af672c938609ead4 : SUCCESS in 7m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5e5fcf20288466e81ad403a65d04b7c : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba0b1fcd7ca410fa6aba02c40595dd9 : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4c94de7311c4ed383874ebcdf21182f : SUCCESS in 57m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17346094411d46f590562f18dd44e923 : SUCCESS in 2h 10m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cac5f908df1e494486800b3c15becbe5 : SUCCESS in 2h 02m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075d447a8b9b47fc9182637fc72f1bb0 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfdcdd3c321f4384a2c310e8e3e06785 : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dac6457cca304e49803bbc70b7af0c63 : RETRY_LIMIT in 6m 40s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7aab52755264ca0a1d3353ee549c38d : RETRY_LIMIT in 9m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1d081b98c5726ee3f9d6cc07d0857bc210a49528","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 16:17:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee3120599cdd8dac74c82a73441d9e4c2cce7446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 19:06:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a91405ac1a644b26adf74f3e4f6622d7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeada72713a64cc381a1ee3e04a68590 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/02b0ede71de4479aa84d3ec8429432e1 : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a1db511b0bf47be85a46762f2e8183a : SUCCESS in 1h 31m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d498a03d2b1346cfb5cb77c4f8db42af : SUCCESS in 1h 27m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/104430849b854631a69a7874492ec593 : SUCCESS in 2h 10m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a7c7aa065b463c92ff19d8e06dc797 : SUCCESS in 2h 32m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5acf79b8cdf4ceaa864a3b18626cecb : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee659470aed544da8d4d10e7aa900de8 : SUCCESS in 3m 53s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/872ccfd3275a489e92eb9b8116210e2a : RETRY_LIMIT in 5m 08s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbc65766033c4b539462f63da96afa5e : RETRY_LIMIT in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"59d6d4adb7a928da85b78ada980f82ce43505e3f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-23 13:43:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d7cd505deff9ee6e0dcfd84acf7c3521879951f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 15:05:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0afbd2062d594649b69be44f3433598a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f46804308b434ba18e726d0b4093da : SUCCESS in 4m 12s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/77241c0618a544b58346522926680935 : SUCCESS in 46m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9fe133f73242475c9c6298555d9f19e6 : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/990c97490b8a41edbf5a776488067520 : SUCCESS in 41m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f5a3536ab0c422c81e7f896982a4f47 : SUCCESS in 1h 11m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca7717dd8d0f48bcb2757765bf0b64da : FAILURE in 1h 13m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8552e1a891b2454e83f8a8c2362a0c46 : SUCCESS in 1h 05m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072e19bbb5784cabb727b1f7946a20f2 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f60bf771e4cc495984fd4c6cbc5e8811 : SUCCESS in 5m 54s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/424f7b703e504007bc7d104260188e7d : RETRY_LIMIT in 3m 55s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e7bcd307e904ad8a02c54e7a607c75c : RETRY_LIMIT in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1cbd4ff8aa13517c0b1aab361098dd768fe9e351","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-01 06:45:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bd9204140e2611a29e9c12c4eaddeeda73bbe68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 08:58:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2679615636da4e1fb161e8fb87f3ded4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24b9b105072740bab1e452ed32539f48 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/85fa9892d3b0458381c04512a6275828 : SUCCESS in 47m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7a4d3ab5757c4fb3bea648deb677c301 : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829f26879dbe416e9faf1754e0523561 : SUCCESS in 1h 27m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae5f8298c3264694ab25181f1ab072b0 : SUCCESS in 1h 25m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8429a4677ece4cc98022929bfbd1b132 : SUCCESS in 1h 09m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a7c479842f14edfbe93e5193321f551 : SUCCESS in 2h 06m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf070d7a50541f58eb2dc82e7e65d01 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe0cc51f78514d65a5444d9898ac6fad : SUCCESS in 2m 32s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ee1272f1ee4f84beb89e1e1a73e196 : RETRY_LIMIT in 4m 57s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ea452f0458c4704bd0d55434264e100 : RETRY_LIMIT in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"42504f2cdfdd8d8f895fcb5867be9b23e0a3a85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-01 10:39:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"446037de8c8db1856c167875ae11acf07ecc16ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 13:49:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/497089806f984f1b8e0a0250492d283e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54896091e414bbb8586b479c4b12b47 : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0c1911336ac841c49d9653ce737e754e : SUCCESS in 43m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f04f089315914e9eb4072cdcb0501d26 : SUCCESS in 1h 32m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d1f331d95fc4bf499be727c63cc93bd : SUCCESS in 40m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fdb0eb4eaaa4e5580d2f75b72c9512a : SUCCESS in 54m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce1a7c6ac47d4642a3748a9d9ee2f516 : SUCCESS in 1h 48m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aca8215cb32049b38515bbd848c7286d : SUCCESS in 2h 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5872b57a29b647feb4f185797d28113a : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cc1c84f811b4de4a6d307ec21daf423 : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43022ecd1a2248cebc061ed3675dfb26 : TIMED_OUT in 3h 02m 39s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/6400d703dacf43f086091c3409070cad : FAILURE in 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"dda93d5f25f63b0c543f0ea1584bc69bf590ad46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-04 11:08:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4d6be3a6836426d522daf1dc0d107d54ad050603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:39:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f789216e26004d35a5042d92976f13b8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b969b218516d42c19504253481f79072 : SUCCESS in 4m 39s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d66bd22e4244229a03dbdc05ca9aa51 : SUCCESS in 41m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/91b3aa45d99144f0a5d2c1ba00f540b6 : SUCCESS in 40m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b187030c2f43483dbcec8bee12cbec68 : SUCCESS in 46m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c67e30df6f434d4a8d17bc84b719feda : SUCCESS in 44m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2792ed98edd14f7f83c9313806796ad0 : SUCCESS in 57m 51s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9c17e082e874f5dbdaf936435f2fd8c : SUCCESS in 1h 58m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bda891846064fdc870ca6172ec7b4d8 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf6d74cc48ca4721ab4dbe139623416b : SUCCESS in 2m 57s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4785c03e937c4d118cee3170c68f959d : FAILURE in 1h 56m 22s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/c2b677bd7fdc451f8088d90ab8c652b9 : FAILURE in 2h 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a62ff268c172875331b793f1516ea085a6400373","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-04 14:00:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3f2a2d5b153dfe7175a94a8de4a188e67aa8e3e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-09-05 14:40:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac30267e7826e76afbde4a039f83edd720d0f906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:42:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f743678ff16c73b95afa4973415a54c730eb385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:19:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b53eab37c254e8c89b03eb915cae2fd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a583125c8d420b83b08a4f316dbf90 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/55cca862b3184fbea6ef1ecf12a018fe : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a36ac20887b4626ae462e57837125cb : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f822224fc9894f25a2c4d59e51f0b506 : SUCCESS in 49m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d27ef3e85d445fa96ee0979a3765105 : SUCCESS in 1h 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af997e3bbebb4db2ada18dabf08e33c9 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9b2bfb04264416aa30e9ee9d7bf70e9 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9c90898620806d5530c162264932b68a765c7192","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:19:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8e096c0ea2d1ae83f4fba37601c138cc4f756b41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:20:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ba861f6f9d4482b771e8f48bf3f071\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d45aa0540148d8911fe65acfab98f7 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef750725f5afb86f81a1e1e7052cc6d61fa47e93","revisions":{"dd0bf97c217b3264009def8995a6b6c53cfd67b5":{"kind":"REWORK","_number":1,"created":"2025-07-10 10:50:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/954595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/954595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/1"}}},"commit":{"parents":[{"commit":"13a30d9831644d0024b6853cc1f26ec59ca1240a","subject":"Merge \"Rename Cinder service name in mapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/13a30d9831644d0024b6853cc1f26ec59ca1240a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-07-10 10:50:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-07-10 10:50:36.000000000","tz":120},"subject":"Disable Peer verification for the proxy","message":"Disable Peer verification for the proxy\n\nProxy module does attempt to verify if service does contain the\nhost name of the peer in destination CNs/SAN [1].\n\nAt the moment we disable the check, as leaving it in place might\nhave an unexpected impact on deployments with user-provided\ncertificates which needs to be re-issued to add management_address\nto CNs.\n\n[1] https://httpd.apache.org/docs/2.4/mod/mod_ssl.html#sslproxycheckpeername\n\nCloses-Bug: #2115487\nChange-Id: Ida475087736984a2f9237d3c7b1a509166d3a623\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/dd0bf97c217b3264009def8995a6b6c53cfd67b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/dd0bf97c217b3264009def8995a6b6c53cfd67b5"}]},"branch":"refs/heads/master"},"798856e691ade15b663b8d186f6d97ef69a3be69":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-01 10:39:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/954595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/954595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/2"}}},"commit":{"parents":[{"commit":"6af76a06cbee08fe15387e2beb5456119c0faa61","subject":"Switch YARN jobs to supported distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/6af76a06cbee08fe15387e2beb5456119c0faa61"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-07-10 10:50:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-01 10:39:50.000000000","tz":0},"subject":"Disable Peer verification for the proxy","message":"Disable Peer verification for the proxy\n\nProxy module does attempt to verify if service does contain the\nhost name of the peer in destination CNs/SAN [1].\n\nAt the moment we disable the check, as leaving it in place might\nhave an unexpected impact on deployments with user-provided\ncertificates which needs to be re-issued to add management_address\nto CNs.\n\n[1] https://httpd.apache.org/docs/2.4/mod/mod_ssl.html#sslproxycheckpeername\n\nCloses-Bug: #2115487\nChange-Id: Ida475087736984a2f9237d3c7b1a509166d3a623\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/798856e691ade15b663b8d186f6d97ef69a3be69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/798856e691ade15b663b8d186f6d97ef69a3be69"}]},"branch":"refs/heads/master","description":"Rebase"},"ef750725f5afb86f81a1e1e7052cc6d61fa47e93":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-04 11:08:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/954595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/954595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/954595/3"}}},"commit":{"parents":[{"commit":"acf461ec7d4a72cf43a2eed0fcaf98cd1292c281","subject":"Switch YARN jobs to supported distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/acf461ec7d4a72cf43a2eed0fcaf98cd1292c281"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-07-10 10:50:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-04 11:08:57.000000000","tz":0},"subject":"Disable Peer verification for the proxy","message":"Disable Peer verification for the proxy\n\nProxy module does attempt to verify if service does contain the\nhost name of the peer in destination CNs/SAN [1].\n\nAt the moment we disable the check, as leaving it in place might\nhave an unexpected impact on deployments with user-provided\ncertificates which needs to be re-issued to add management_address\nto CNs.\n\n[1] https://httpd.apache.org/docs/2.4/mod/mod_ssl.html#sslproxycheckpeername\n\nCloses-Bug: #2115487\nChange-Id: Ida475087736984a2f9237d3c7b1a509166d3a623\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/ef750725f5afb86f81a1e1e7052cc6d61fa47e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/ef750725f5afb86f81a1e1e7052cc6d61fa47e93"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
