)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~971802","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~I5c6e0f57007dd7c99e42439a0a1daf551f15bdb9","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-23 15:25:32.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-23 15:25:32.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-23 15:25:32.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-12 14:44:02.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-12 12:51:46.000000000","reason":"removed on reply"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-23 13:36:01.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5c6e0f57007dd7c99e42439a0a1daf551f15bdb9","subject":"Enable using of pymysql","status":"MERGED","created":"2025-12-25 17:10:45.000000000","updated":"2026-01-23 15:26:37.000000000","submitted":"2026-01-23 15:25:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971802","meta_rev_id":"0e4fe2be6927cc00ee94c4fd7f1dec205bf0efa7","_number":971802,"virtual_id_number":971802,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 15:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-12 14:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-12 12:51:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-01-23 13:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-01-23 13:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-12 14:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-25 17:36:46.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-25 19:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-26 01:11:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-26 01:11:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-12-26 01:11:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-22 13:30:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"4c4bcfe041320ee305022d1b7feb19d81a14cd67","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-25 17:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561c4e964cf0e3418dbc37fa16377fbccbc1fb34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-25 17:11:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"95aed5ac0c264cbaf90908ee5e4fa349364b514a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-25 17:11:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3973c43332e06f893175477a4989ce6aa13c0a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-25 19:47:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681d89867fff4b87ae2ec8b32b8dc1d8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73da73e3a8264f6ea2535e04f2889217 : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c8febe31b48445a5a5cd7fd9cdfe65f2 : SUCCESS in 1h 00m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dfb1a45d5ca8445f834a1a0c1b3c52b5 : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/71938061603b444ab52e714dd07e7a02 : SUCCESS in 46m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65c0774da21d40d5b7feef168ef89e9e : SUCCESS in 1h 03m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8982ac31490447a8bd44cf74dd3d8a1e : SUCCESS in 1h 33m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/585f6b5f88a54aa99f43c92731bf9b7a : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac2a85442c514bdf833d5cb47661abe2 : SUCCESS in 42m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ba240563afbc4afbbc53330c9724761d : SUCCESS in 1h 36m 31s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a4539a721a814fdd97b764e01dba45bf : SUCCESS in 1h 31m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/227bd9939bcb406893fdea05a4cfbf14 : SUCCESS in 1h 35m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4570ff3927547988e9831aa16db58da : SUCCESS in 2h 12m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b970fc2e4724634bbd67a0c7a14f53c : SUCCESS in 2h 05m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0013244b6bb441096bb055958871334 : SUCCESS in 4m 17s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89d0c56f766347d49d6ec1ae6339e79e : SUCCESS in 2h 28m 42s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/649e6e1a3b424723aafe04b013d5da20 : FAILURE in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ed4805d878a80631cdfbc3c248f8d3cc0461be6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-26 01:11:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bae94dc2d9d381eff56e2f1d02076ed2f85b3636","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-12 10:06:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"243df8179f11d63cf8ed92f745b5ba0987983736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 12:50:11.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"856688558ce754df6308be12bb258865e5200a93","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 12:51:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dda3c56355d76880414baaf72431ccfd44dccdd8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 12:53:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4b68c9fb8eac032100281c71a27a3e73e6a310b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-12 14:44:02.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4157635bcfcd886c93cd5856207c73555231891b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:24:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca76bf8dc7d142e2bdb6581b68f10f53\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6ff9d50d9e64b37962bce4a1f85df4e : SUCCESS in 8m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1c215da9fbf144b89b181e68547c1c0f : FAILURE in 24m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5c71dec69594c2bba5301bc27119b00 : FAILURE in 28m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/63c3bedf698f4f65aaf6ea55505885b0 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d59bcda50e3f44b380acdf780be16c78 : FAILURE in 25m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d4bc24c719424df993a4783b7d794dde : FAILURE in 20m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ccfd3234a774f9c8973bd05a905daec : SUCCESS in 1h 13m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e3c4f19af5d4961865fab1bf0454fda : SUCCESS in 1h 29m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fba22d58b5eb40f5958f0ac8d6b9599d : FAILURE in 16m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0320d2d51e1a4aa487c95c7dcd6261e9 : FAILURE in 12m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ae8eb3edbdd4c219e0fa711e3d3c3e6 : SUCCESS in 1h 10m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/007f5fdb777f4f88b6ea8ee7cc677993 : SUCCESS in 1h 07m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d6cae15519e40af9960215b5868248d : SUCCESS in 2h 18m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567fa3a373464e9aa0a44a883ffea919 : SUCCESS in 4m 51s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6febb06b5ac24724a21d846f86068b6c : FAILURE in 2h 28m 04s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/edd62e1132b6459cbcb81c25dd347f2e : FAILURE in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc8d11043e61f08dcbb71cd204b4f06ae7d3f400","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 10:00:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b63444604092de5b21ac4ac93d0c1bb193482f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:24:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea239f7881c34dfb90d82afc043ad2cf\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0518935758fe48ce81dbb44d06bd72e3 : SUCCESS in 9m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f7e270ae71574a2291ce5e63ef698873 : FAILURE in 19m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb8fc4f010b44bbb8063dbcae1f2875c : SUCCESS in 1h 16m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e27b8f2b0a7049c7987de0241fa85920 : SUCCESS in 48m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f4c6286f4414acead37b657edd27e4a : FAILURE in 26m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b3fa358419e0498f9eab2ded56e40dce : FAILURE in 16m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9474e59eafd4dea9164539e0c0dff67 : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2fdfdcc10d74004b2646e519ff2de60 : SUCCESS in 1h 28m 35s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/096568093be54534b6d9705205fce7c9 : FAILURE in 13m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6c7242a060c46a7bdd77ecb3debed45 : FAILURE in 18m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b22f07e9008045acb45ffd4f3ffa6996 : SUCCESS in 2h 07m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ce8955da82471789cdb63e958a27c6 : SUCCESS in 2h 17m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/702eff7e796745da82a2c151ab966f4e : SUCCESS in 2h 17m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e84a7a275c4ec7b92b34acb8d88f7c : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78c1a7ac6cc24de3b54cac00a4a1e4b4 : SUCCESS in 2h 15m 33s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae01de0858fb45cf897bf2310b6ee065 : FAILURE in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ded904bd70368326a55fe54ddc88059b471ee2dc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 20:38:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"913ec2bee1335ba24eba88e5f90467544eedf8d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 22:54:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8236994a96954bc6bc0c0290d70f6cdd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519c820b37d141ea91283f257331a446 : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e0aa49bec0904ec1a2ae0e02cbec7f00 : FAILURE in 12m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ce8c6ba625d4449a9e602a77615a001 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3d2d74051c304464bed246ed9dca4460 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ccdde5b912c04ac39f22343aa98ae1fa : FAILURE in 21m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8ac556479a3432ebdea8f5548ec5a3f : FAILURE in 20m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f35e16ea6814d10b77c10bc4df1d970 : SUCCESS in 1h 28m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74dbeecefae4450284899260be503617 : SUCCESS in 1h 29m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/03c12cadab03465a85c0503ec67d537c : FAILURE in 12m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bdc33ab723e045ea9f22eb17626d4abd : FAILURE in 13m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5830b41f4469437a99ed7f9c5cef0914 : SUCCESS in 2h 05m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a776a9c62984ad39692810cca9235db : SUCCESS in 1h 01m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/064d7ac49fbd40eab7340d15b2c7c2a6 : SUCCESS in 1h 05m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201ce628e5c747dcbc13c3e90ae26eca : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72568dcabcd848d7b78738f671b16c24 : FAILURE in 2h 12m 19s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/fdd5dc1751aa45a68ee24fc808581321 : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f855166fe8fe8c1e7041aef29b6e254caf665f3d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 10:33:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65f5ccc626b95e076c74a36962e8676b9f9b176b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:09:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b96edb39f49b436fa432a12d9e0e14ae\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6c3a87f40084d8998d448f76a7cc141 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4baa677148274d49883e4265d26da6d2 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/de3adcf6669a4b3082f59517967f3b30 : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/12a95ecb2eca435c8fc4a38f82c7633c : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46e99cd194934fc9bc735136fc508c2a : SUCCESS in 45m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/319ce170637f409ebdc343a1828b0ea3 : SUCCESS in 45m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b04df2c10030435090499c25a8938ce5 : SUCCESS in 1h 28m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a8be8dee80f4177a69abc94c041ce77 : SUCCESS in 1h 24m 35s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/77860b0edba74ef2a81be1eb54ae2ddb : SUCCESS in 1h 40m 16s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c86630f8f86f4e09ae88dd89e799ac46 : SUCCESS in 2h 17m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1dd9b94e14e46b99b39c3881d51b08b : SUCCESS in 2h 16m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ce9777d03e54034bf995459db8300ae : SUCCESS in 1h 41m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/097237dd83eb49d7bdefb0531540e645 : SUCCESS in 2h 22m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338293b005114356a494732794258773 : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac10dfe012e541c0b74e86c82a439963 : SUCCESS in 2h 29m 13s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/d49bbc24d67b42f2bd88afd5b01762e1 : FAILURE in 1h 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dda23a2ab1bf51b33a3406c67427e23a1cb3dcd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-23 13:36:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7fb6cce4c1b3d2d953b1cf44b526dc89fe498261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 13:37:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"521416bfc513f2884bf323492bab366467bc9eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:25:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b8a8b67518b432999c102f6ff54a39a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7a7468658c6485f9a9491a1ddf88ab1 : SUCCESS in 11m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/110a850f90a4400e8715193586e0a3da : SUCCESS in 46m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/972f8f954de5418684cb6aa05337d0f6 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1921a0f981bb45c1829a324463319ce5 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4bf190c0e03d4fd5b90e83ac512cbe56 : SUCCESS in 1h 38m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/332eaae5f2344c95b81a38aa0cb3106e : SUCCESS in 46m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c2849e8f1974044811204b7a7351686 : SUCCESS in 1h 35m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/766bdbff8981456886ae05a39a2b32fc : SUCCESS in 1h 29m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82aa8e118d845388a593d55cfc2176d : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"98e3fa870ad1be448a5f1d7478f08e04abfbd18b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:25:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0e4fe2be6927cc00ee94c4fd7f1dec205bf0efa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:26:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde5c77b037e4ded82628443eaa24f81\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1317342b8646c5aeb3c8b1f603a5b9 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6b72601280500ef45c258c834e4b89ab034e3f8","revisions":{"20f017338a2173a115f2456464c3e15f227646bf":{"kind":"REWORK","_number":1,"created":"2025-12-25 17:10:45.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/02/971802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/02/971802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/1"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-25 17:10:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-25 17:10:45.000000000","tz":0},"subject":"Enable using of pymysql","message":"Enable using of pymysql\n\nChange-Id: I5c6e0f57007dd7c99e42439a0a1daf551f15bdb9\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/20f017338a2173a115f2456464c3e15f227646bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/20f017338a2173a115f2456464c3e15f227646bf"}]},"branch":"refs/heads/master"},"d288c6c23f4ef2b0ef8bd6c392e49bfc92711d88":{"kind":"REWORK","_number":2,"created":"2025-12-25 17:11:26.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/02/971802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/02/971802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/2"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-25 17:10:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-25 17:11:21.000000000","tz":0},"subject":"Enable using of pymysql","message":"Enable using of pymysql\n\nChange-Id: I5c6e0f57007dd7c99e42439a0a1daf551f15bdb9\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d288c6c23f4ef2b0ef8bd6c392e49bfc92711d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d288c6c23f4ef2b0ef8bd6c392e49bfc92711d88"}]},"branch":"refs/heads/master"},"b6b72601280500ef45c258c834e4b89ab034e3f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-12 12:50:11.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/02/971802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/02/971802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/02/971802/3"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-25 17:10:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-12 12:50:11.000000000","tz":0},"subject":"Enable using of pymysql","message":"Enable using of pymysql\n\nWithout pymysql, SQLAlchemy tries to load MySQLdb and fails.\nUse pymysql to provide a compatible driver and fix startup.\n\nChange-Id: I5c6e0f57007dd7c99e42439a0a1daf551f15bdb9\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/b6b72601280500ef45c258c834e4b89ab034e3f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/b6b72601280500ef45c258c834e4b89ab034e3f8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
