)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~971808","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~Ic268e75b832638dba8c980f9d266f0ba019d0de4","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-30 14:22:02.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-23 02:19:56.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-22 22:04:55.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ic268e75b832638dba8c980f9d266f0ba019d0de4","subject":"Remove panko roles","status":"MERGED","created":"2025-12-26 01:21:21.000000000","updated":"2026-01-23 02:21:44.000000000","submitted":"2026-01-23 02:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971808","meta_rev_id":"d64c9acdbd15b52fc37d88edd4cf8079ae1465d1","_number":971808,"virtual_id_number":971808,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 02:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-12-30 14:22:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-16 09:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-12 10:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-16 09:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-26 01:23:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-26 01:23:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-26 03:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 10:04:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-16 09:38:59.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"4e449e84655e886eecc816cac726625eefbf315a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-26 01:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f3160a8c08818d92bc0b57fae49cd7d702a0914","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-26 01:22:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7f899b6c8ecfc1ef27908344e4418ad69f7ffca9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-26 01:22:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0e9114d0668ade7c31c4f929b3202991f6ba7079","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-26 01:23:04.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"ac0c49878477aaaafdd333508663d4fbe6b321a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-26 03:37:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b8c3fb632143c5898d192b655e1bee\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c80cb9d8332e4caa93b4b61238f8ea27 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9b1f82a52dc5440d87fbce83f7a9da9e : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f4e86798a05f49efac428353f33e85a8 : SUCCESS in 1h 05m 02s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/484cf75197e4454c92738dd18af3ba60 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c844882dd28f47feb34e14a87dddb63a : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b936832fc42444ffa31853338c07e538 : SUCCESS in 1h 32m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fca2367c18bc4d7d8437b607962a2a5b : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ace36af92b46f8bbbf6ae5e5ef47a2 : SUCCESS in 1h 09m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf16478832ca4e8298ec3131dcd3077f : SUCCESS in 58m 31s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae49bec4640b47db87131f0a43aefdfe : SUCCESS in 1h 41m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c1ba959bdaa44e692c60d09c8a6f242 : SUCCESS in 57m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/655e0c6b6b224657a50dae092dfa35a7 : SUCCESS in 2h 07m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67f02d95ae0c4021a3b8bb4867ca8738 : SUCCESS in 1h 44m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d113814c86fe4681b0c64bc5a819f964 : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d89064132e5a438fb80ee0b6f7655467 : SUCCESS in 2h 04m 05s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/5de0ec0c80644a1aa48c4b4ce28a0f64 : FAILURE in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b24d94d7e9ca94e16957abcff1c157b6ae306a2c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-30 14:22:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5b7d3b7d2b2869f7f7b2834f29e526a063e0f03b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-12 10:04:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ea2c0d6b0f681465e52c6a68e7a6ea13431f66c1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-16 09:38:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"92e04ebd6b783d2048e7516a222ea8c6b84c5af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 09:40:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7f84c637486de23ac33be85bf0c1d171e366f460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:11:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33ef7c0742a54567a9b02b0064376cc6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3a8edb9722b4ba9a4f2d9efafe06bbb : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d34b09bdb02b4e66a433f2ff18992b78 : FAILURE in 24m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c447f74a1aba44598322c10ec41da6da : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b5b7580ec35242c7b5d67805f5e54242 : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/94a9a0a44b7740019a3e7c259b9aa40a : FAILURE in 19m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3518d37b6f88424c80767a9f5af13f56 : FAILURE in 24m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9de3c6c7b0d4d0989b897d7f93cc386 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ac1fc87579a451ab8840eb7a925fcaa : SUCCESS in 1h 25m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a470c518b5846f69b8807558e3eebc4 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"1d1e8947d35f5978d4ec23932513337b40b8d530","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:22:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2721575fc33efa98d042df83628a5958abbde87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:07:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc8395442c05446596b70731eab89494\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4412b16c04604a058438c75c31309e94 : SUCCESS in 11m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a18a59ee2add477d9ad57b1f796291ad : SUCCESS in 53m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/02f5ecc7bef24041bcf5149d2a026214 : SUCCESS in 1h 22m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c05bb3e6a0094ca2b3d137c2a5809b09 : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f7775a4051c4aeb864363e8bf7b6e5f : SUCCESS in 53m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/540cb8729ec84fe4923c8c7c7ff3f7ce : SUCCESS in 1h 41m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5a235392f13441198ae82dacccded66 : SUCCESS in 54m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8467034f6d3c49268093e1b70f7b0be6 : SUCCESS in 1h 38m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e08d0c63f474903b15008b95693b9bd : SUCCESS in 1h 16m 50s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a171d818d021490a8f21fa6e2b84e6c5 : SUCCESS in 1h 50m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339a3e05fb354c70950ed25b5273f836 : SUCCESS in 1h 02m 40s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a85812f2b34d4ec8ade4da08409d8ff6 : FAILURE in 1h 10m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eea41043aa444c8bba05b83a2f8fd60 : SUCCESS in 1h 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3706ef21e6e14d96867c3766908429be : SUCCESS in 7m 19s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/545a1e976b784618a8a4dd39d292bb7c : SUCCESS in 2h 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/f0bf40d3f9934e2cb4f04259d344704f : FAILURE in 1h 08m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d47ae70c31059d09f7ca4e5d2c7f6415e09c230c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 15:53:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f14a59055c805528946694d7ea154d12182164ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 18:20:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c88b5d776ad46aead1b97dccf9a94e1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/295e92e0c4924e8bbcbf1b18b80f1237 : SUCCESS in 13m 37s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/dd666951918242258fc21738e7a70ca6 : FAILURE in 19m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f4db4848d21a4b8fbf0a5c450e40cec6 : SUCCESS in 1h 32m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/07decbf1687842e5ac188195b4b67848 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3de5920d0f474b16a9e94af61442ba6e : SUCCESS in 55m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e1fe31c73974bbfb19d7fec2a33721e : SUCCESS in 1h 45m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f72b9fc7af41f6938180bfbf3f9a74 : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f24f0f7807c4415692b7b86c80727a91 : SUCCESS in 1h 18m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/96f6eced23c6422985e0d303f13083a2 : SUCCESS in 2h 20m 17s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eed9deb3794544d2833b6e6ce764838f : SUCCESS in 1h 09m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d51f8a00efa34e6c82cbecb8a51e589f : SUCCESS in 2h 18m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d7304e016fa4601a0215b74945d5997 : SUCCESS in 2h 11m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/755c4c0625204988a51d05d1ca95d71f : SUCCESS in 2h 20m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812c1a2aa7d047aaa474795f40c9583f : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20f90de29922422bba9a5b5f11b94aa6 : SUCCESS in 1h 23m 51s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/4e8382ce3ee34b98b8d38c0d3a75ccb2 : SUCCESS in 2h 06m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1815d1eb9ddcdd9e2845bb8afc60fd580a1240f3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 22:04:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a9e850ee15b0a4a8ecda72252812e93c7664bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 00:34:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45ae714625044c9b15fe82ae064e768\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1897404535014f539751e7cfb4535bdc : SUCCESS in 8m 44s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/443961f26b454fc5b77ffd236ae9125f : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/11b10ffd823247d3870d5b44f82a8ee3 : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cdba802a762a400cbf91bb857278985b : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91a8a6c909f74ad4a3f25aa26edc67a1 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7832985fee39409d90060442f1cce08a : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b44de1a39c84300832d33774982e1f4 : SUCCESS in 1h 30m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/953628fc411b4c7e8518b23049e0a30e : SUCCESS in 1h 35m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e6b965cab2e6498f8dba9861642b4f77 : SUCCESS in 1h 05m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc7fa56796474167971f04c81aa6c354 : SUCCESS in 1h 05m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7711f0b7baf14c5280caad88b9735a7d : SUCCESS in 57m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0973ab0e8cb846f39d70e843b89b80e0 : SUCCESS in 1h 41m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/122f6e60f0cc44518c22698d2eac7d38 : SUCCESS in 2h 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b545d4bc4a4560bc56714e15da3334 : SUCCESS in 3m 28s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa28bea5138e4bcc86ccda5d0b7bafe3 : SUCCESS in 2h 22m 34s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/9214bd205cca4cc18c3f99f804a353bf : FAILURE in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10528e4f235bba7940181d1f614e521268b68a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 00:36:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b5f7294819cbfce6e67f0138238934f7bcd0887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 02:19:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7ca3294600c4a0593d53a9c9b32cd94\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4ebc14704074987b974dc07955e3f5d : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a7523206d09c4ef89aed9dfbc54c0526 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/06e889f1db8e45919cf993e1e9f33cf4 : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f832f043bb3a42028903f17e2f5424d3 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ca6831bb902f4148a1571e537ab7a367 : SUCCESS in 46m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b693da538c904c6c93118a85ffca2744 : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52e91b92d7ce4fca9fd753b87e27520a : SUCCESS in 1h 38m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6486fbca11244388dd553f4aef5777d : SUCCESS in 1h 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95cabecc853406b84fb38167144d04e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"83af9efa91550ac6d13b1c5d36e906d32f948306","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 02:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d64c9acdbd15b52fc37d88edd4cf8079ae1465d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 02:21:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b272b51be084c2a8b834437b8332513\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5be6eae6274043b740403d66a3c4c2 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8951e2eaecedb7231a0b18b0405cfa1bc79a1c54","revisions":{"1583402bac8b72fdc9d5944f8d1aac161b1c35aa":{"kind":"REWORK","_number":1,"created":"2025-12-26 01:21:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/971808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/08/971808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/1"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:21:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:21:21.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic268e75b832638dba8c980f9d266f0ba019d0de4\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/1583402bac8b72fdc9d5944f8d1aac161b1c35aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/1583402bac8b72fdc9d5944f8d1aac161b1c35aa"}]},"branch":"refs/heads/master"},"71a46eb0871d209d62c8dd8e6a0613ecec1e34c5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-26 01:22:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/971808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/08/971808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/2"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:21:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:22:07.000000000","tz":0},"subject":"Remove panko roles","message":"Remove panko roles\n\npanko retired some time ago.\n\nChange-Id: Ic268e75b832638dba8c980f9d266f0ba019d0de4\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/71a46eb0871d209d62c8dd8e6a0613ecec1e34c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/71a46eb0871d209d62c8dd8e6a0613ecec1e34c5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8951e2eaecedb7231a0b18b0405cfa1bc79a1c54":{"kind":"REWORK","_number":3,"created":"2025-12-26 01:22:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/971808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/08/971808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/08/971808/3"}}},"commit":{"parents":[{"commit":"e1412255933fc9d50a432ad334747b5b41c54e29","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e1412255933fc9d50a432ad334747b5b41c54e29"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:21:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-26 01:22:52.000000000","tz":0},"subject":"Remove panko roles","message":"Remove panko roles\n\npanko retired some time ago.\n\nChange-Id: Ic268e75b832638dba8c980f9d266f0ba019d0de4\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/8951e2eaecedb7231a0b18b0405cfa1bc79a1c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/8951e2eaecedb7231a0b18b0405cfa1bc79a1c54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
