)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~981149","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~I387c99010a1daa58d3a968d74b2bf0956146ebc9","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-05-15 22:49:45.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-15 18:19:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-09 20:10:24.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I387c99010a1daa58d3a968d74b2bf0956146ebc9","subject":"Fix skyline yarn-based installation","status":"MERGED","created":"2026-03-18 16:27:12.000000000","updated":"2026-05-15 22:51:29.000000000","submitted":"2026-05-15 22:49:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981149","meta_rev_id":"0470c84c40755acdbbba64cdd00cc991d09a22a0","_number":981149,"virtual_id_number":981149,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 22:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-05-14 11:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-14 11:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-05-15 18:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 19:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-18 19:27:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-18 19:27:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-05-14 11:41:26.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"8d829ae4e22f15e3707df09cbba11182c5aa7928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-18 16:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bf87e4f0c3d0718d9ce2582c2af1531b9e0d6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-18 16:29:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7319094ae00abc02f76b6fac47af39382491c81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:10:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd874c97fb1e49daa536029e135571f1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6079424675d4eb5930c8b8c18cd406e : SUCCESS in 12m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d73ce797ece24eeea25b2ed28747261f : FAILURE in 51m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/608bb59209d24df38cb7fb66a2243128 : FAILURE in 52m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d84f227df4194e538b10ed3579a34457 : FAILURE in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7120ad9c73354f8ebba3b9e223153951 : FAILURE in 53m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/89a8f7f6d4e445548f305c6ae7c27205 : FAILURE in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdcc4c9c316145f1bb7a81dde00ee1fb : FAILURE in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fa5f7a8689c43cc8e1bf21701ab21c1 : FAILURE in 1h 20m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c57f037c86d421f84190c30e94277af : FAILURE in 1h 59m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d188d0738266464d8cf6055b4ebe395c : FAILURE in 2h 22m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f0d2fa3f464b049cbd37fde3cff7ae : FAILURE in 1h 48m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9fdd836d6424615a1287595775437a8 : FAILURE in 2h 26m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006f8b1340e5425983475b1ae609dc0e : FAILURE in 1h 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ba8cadb4fa46a7b98ea8db4227f796 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b55df17052b14e2586d0fd38912cbece : SUCCESS in 2m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e97c4e5fa2247d8bd431d0b943986d5 : FAILURE in 1h 07m 34s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/19955700010e4387ad167b29f9f66e55 : FAILURE in 2h 18m 03s","accounts_in_message":[],"_revision_number":2},{"id":"67f84a24dc1805857c4c261b14241ae81d7d0590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-18 20:38:19.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"625a52a5c2979c91de5dc9aa673bebedceaee9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 23:14:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/982e72eb8d5049eb94a98039855867d2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d7be97d436f4e00a073eec3e825824b : SUCCESS in 10m 30s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b827c1d9857c4d8982a2bc27e20f62e0 : FAILURE in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ad717599fec441b9326f9e51ea4d05a : FAILURE in 1h 32m 14s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/926c404829534c33b4ea495fef5bde1e : FAILURE in 1h 02m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/09779c52a1154fa8874c3ace3e3438b0 : FAILURE in 1h 11m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1117f23f03364550979173b5fa289a10 : FAILURE in 53m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a12ed42b5863490e9c51f37135daa277 : FAILURE in 1h 14m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b886dec3f1884c738e6e3a5d2761559f : FAILURE in 1h 15m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b28fe0474b84eb98dadfc8ccf368af5 : FAILURE in 1h 11m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbadb9f7c0db4fc0a526f481542bcdb2 : FAILURE in 1h 38m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d55ffec38f9849b8bf5b1c88d2f5202a : FAILURE in 2h 15m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b38866116da4d3e80870a4d74b9f536 : FAILURE in 1h 08m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd2c3c5044a4032b21e59555e7cf6cb : FAILURE in 2h 29m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a36b4071039b4da1bb7b70d0ad9f747e : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5465059f21dc4469b79ae750387ed568 : SUCCESS in 2m 30s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3731f9ab1bf6429f9772ea3dd83cf764 : FAILURE in 1h 20m 29s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/7673f3e20f794365b469bc82fb8d648f : FAILURE in 1h 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f2ceb51eaa3d13d6684cb5338c94fa888e699416","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 15:22:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad29759c9a78586a3f9f23e4672994286a5b2fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:02:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9f66adc71824c15bcd7588c68d8ac13\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f689b486518a492982205c9bcf9b1a53 : SUCCESS in 13m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/42d15c0fa3844989b49f63c65d5b054e : FAILURE in 59m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5a73be1c18674bae8f3b9f90674492cc : FAILURE in 1h 30m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/73bd5cc03824411ab7608eb49ad2bf0c : FAILURE in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c7d4f7d1c7ae4ce9a187cb81b16e0d3b : FAILURE in 1h 32m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da5537e62d544d7492c94979954be15f : FAILURE in 1h 33m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d43c4ccdcc145db839197b7e0880ca6 : SUCCESS in 1h 56m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15bf802da8344cacb4f8019674c174f0 : SUCCESS in 1h 55m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db4ba0be7a2c4e4b96b53b602703b5b2 : FAILURE in 1h 59m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b830cc6cbcd47b7b25a0dc2aa1e373f : FAILURE in 2h 44m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a709b0ea5c94d0eb5acf0c0a224a8ed : SUCCESS in 1h 31m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84bfbf0b60044da98a0c1131cb4d9c94 : SUCCESS in 2h 12m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204e310da3324401bd9b24aa1d3f7d3a : SUCCESS in 1h 28m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83126474d19444508b9153579fad9628 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/560ba12ac2b94a20bf46f1025618ccd5 : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/424f917626654f09acc6f5ef3c34aeef : TIMED_OUT in 3h 09m 55s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/a4fe885e3dfe4c7a8f4f494829658c15 : FAILURE in 1h 33m 20s","accounts_in_message":[],"_revision_number":3},{"id":"b29f01d91593777b1788800fda5aa221c10233ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 11:17:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09ae6e49bfc60e2a56b14eb4901f17f06df6cbd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 14:00:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9b613549cf34b7a86dfde3e585be5b4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28488f522cbf44f2addb00f05bd8365e : SUCCESS in 12m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8822af599c8e46f5b780dbed37cf6ba8 : FAILURE in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e374b929416d495188955656e53066c3 : SUCCESS in 1h 45m 26s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5a48c9573e904626bde9bb827c504c41 : SUCCESS in 1h 00m 39s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e55b5d3f3ced48d28e0ac1751934a4fd : FAILURE in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/776804fd40054b1e92568a9d7c904e62 : SUCCESS in 1h 03m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cb35bb2b1864f64ac21564c6b51b1e1 : SUCCESS in 1h 52m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cf907024ce34816aa8f52d02aee582c : FAILURE in 1h 32m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2842c9e9ad2425397902a1fe1cd16be : FAILURE in 1h 45m 02s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d2629209b8024cfdb472eb7081c08d14 : SUCCESS in 1h 14m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0e3f26b61b4cb0b0e07a7add8a51c6 : SUCCESS in 1h 24m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c72ce0ece00c4a2abc0c47f93627ce75 : SUCCESS in 1h 11m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92e0a954f015473f8318286f0ea41fb6 : SUCCESS in 2h 35m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40316d6a5dd145a2932ce95869841519 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a16c0d434a478694f0270f9eaa7af6 : SUCCESS in 3m 33s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb069d0794024f5ea183a2b1fd9ab9dc : SUCCESS in 1h 56m 34s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/f29b04b4121246fcbfe64276929bd13e : FAILURE in 1h 19m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8ae317f90fc222d0f914f24fc5d7e68ea015b2f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 08:00:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e8e3296126e9b88bf607a6651e275a3cca7a937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:05:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65dc0641fd3c4f6e9aa334c0a3c825d3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1415e8284a9d42beaaac8fdbfd0d4b3f : SUCCESS in 12m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/265d1b1bd12a49b48537b65f195464b3 : SUCCESS in 56m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1ce135ca82b24c7fb9239544208ebe0c : FAILURE in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/39f6dea728784f84917396479df3e993 : FAILURE in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/51da14e373de41d2bbacca3c3d5ae007 : SUCCESS in 1h 02m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fd003468f80745718dac36908a7d075e : FAILURE in 54m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7959d2508aa498dacd0e2ae730971dc : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fef0347ca5564d8081c9d8bd98801156 : FAILURE in 1h 36m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8c9d6e5434c3496594c7efdd0c95ea9e : FAILURE in 1h 18m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f6f9ea35c4d48e695ead4ff88da8975 : FAILURE in 1h 55m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4d8781aeb9940b99beaa3a923e6dcf3 : SUCCESS in 2h 31m 00s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233302b8a4cc423e928d26c61f73b973 : SUCCESS in 2h 05m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b208065666f4c55bffe895382387d39 : SUCCESS in 2h 43m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896140f88e064135b3ffd32e183830d7 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/306e2e907944426cbdd58afac256dd66 : SUCCESS in 3m 05s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ae8864676254cd1a4f7624d069a75a9 : SUCCESS in 2h 55m 42s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d49fcb428554d7792ca39ac7d2a74e6 : FAILURE in 2h 04m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f085f5b5f1f0dd7ac6397ca79aa5b3e887d55840","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 20:10:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ad9ccccc9b070a84af991dfd387353032e82407","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 21:25:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c111e1655ce0761599678ab224cb859644c396f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 23:10:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3044a7d944c140d591ec68a0083202c7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0bee539f7214b8aacb9e15659aea34c : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3719a7096c4646cb89a452e2423a9af7 : FAILURE in 47m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/16ce3951d46242a1955d815c2f76bffb : SUCCESS in 54m 24s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/062c2c00ca214986aea0f50e8e34eaf3 : FAILURE in 1h 17m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4594b105eaad4679ab3504c24d0b075a : SUCCESS in 1h 24m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bd8f11394334498ba4e83e4312118e3 : FAILURE in 1h 28m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54f8c9cfe5774916bd8e6a728453ce8c : SUCCESS in 1h 42m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8053e81860b74dd0bb666b49a54d0752 : FAILURE in 1h 49m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5bff22365b87406aa30d219187efdadd : FAILURE in 1h 15m 13s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b25e5f9e3c748abb8638d6be1df5bdd : FAILURE in 2h 16m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2e01666370d4d45b1bd63a4aa6084a0 : SUCCESS in 2h 26m 35s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e923895e12f946a493b8448371649fa9 : SUCCESS in 2h 30m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b03c611cfd04c86b3081a18aee4e0df : SUCCESS in 2h 50m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc9e1f8e9064ead90cf87a62dc0d6ac : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff0dc32b9e4643769cce3e84ac9f830b : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3704513fdbc146028d13aeb4ba02f97c : SUCCESS in 2h 42m 59s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/27fbd7250ba9438c8a2fc0c583ad0827 : FAILURE in 1h 42m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e718d79705bdb49e10b1366e7092d8c14d89bb1f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 18:36:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e913cb9292bf3c8fff9e21d6a09b5639c42c0879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 21:20:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ef2c2b06b604d4cb66430623aa9ba09\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97dec751d3a045e898cd3be914c1fb28 : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/65c7c2c7563f467aaea001f8bc80359d : FAILURE in 1h 18m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2fb911ae8dea46e68be0bde262ad3f4e : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c2f18ea555eb424fac8d865fff5f0181 : SUCCESS in 58m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ab21382a61b46ecb0b0a40a9f670dc1 : SUCCESS in 58m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b44792839a62481cb301762d663be215 : FAILURE in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ff877895c7945b88c7505e0b2fadaee : SUCCESS in 1h 44m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/759075270c2d42078a4e778dd75b9cb2 : SUCCESS in 1h 16m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f6b686e27284fd786aa9db03d9a04ac : FAILURE in 1h 13m 51s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3479b35ff4574a0881d065617e317912 : FAILURE in 1h 35m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b958e2770a34a15942d782177999a5d : SUCCESS in 2h 31m 13s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7a2ca14beef48c6b017343fe891026c : SUCCESS in 2h 26m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77b8d5f41e50431bbe254462fc55540d : SUCCESS in 2h 35m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4862a33f81485f99f472a0e8822b0c : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436bc5e7f70a4d4d84a98731b9c85314 : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/181a1140e9e3406d8706087c8886cc1a : FAILURE in 2h 16m 21s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/36de19c34f7d4a8baf074f34f495433f : SUCCESS in 2h 17m 45s","accounts_in_message":[],"_revision_number":3},{"id":"20bfb8e8727802958ee62aac18ed2658e91761f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 09:04:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bc244bbc0db8ef3b44f8e4751a8bbc03a2e08c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:01:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b1e2794dfaa429a9676281177bc88d0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a2d71248a3147cd972b1de7ff88b581 : SUCCESS in 11m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3c7bd3391301411489c2f7bdf6ecf5cd : FAILURE in 1h 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e0f8325b457548cb9e785acf16141909 : FAILURE in 1h 30m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7d13c8177a2743129fabfc8526e66d03 : FAILURE in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d3fec2bd4d6444da0dd0af214770e67 : FAILURE in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8be0c8f3cef4d438a11a96c71814371 : FAILURE in 54m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08e98e1be404040b1bd5cea94e8b307 : SUCCESS in 1h 41m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c20b2e8553a048a99a0f97c10b6546ff : SUCCESS in 1h 53m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d52a95e47b24926be7d9bcf1568da73 : FAILURE in 1h 07m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0d450e332df4b1f986136d91bc6feb0 : FAILURE in 1h 47m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71ce7ddf0c104084be93ca91c8d0b05d : SUCCESS in 2h 28m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a0d88e5af3a49b98e2d56989aacb3c7 : SUCCESS in 2h 35m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28c7bd1d60cc4047ae38f1f8d855df63 : SUCCESS in 2h 47m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34812c72a53849468d654060648e5e9d : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec65f6d4d7624daab3d54baf34c583df : SUCCESS in 2m 48s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c407b06432a435ba575fe42e18b45a8 : SUCCESS in 2h 31m 29s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/3dfc9627847243658f08e9dc9fd74462 : SUCCESS in 1h 36m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d56580645f369eab9e41f27507f39530876b855e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 12:19:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9430bc5be8e57250daf14bf0beefb692b71f4373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:01:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46967a66b9e94d34803d4c78b336ffcb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfcd0b3bee3748fc8b7e672c9172b173 : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/22a3ca2e7c5e46fc89e0828ddb94a418 : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/80a316cf8c12400abe175a5d363bc811 : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/067d10eff4fe40a3837d4344a97f581f : SUCCESS in 1h 41m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b113e50b47be46d6b39806daf3ae4510 : FAILURE in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/435a2f3d6cc443a2b9a59fb68bdc98f3 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbbd93e1c2074840bda9c9024fe2fdcd : SUCCESS in 2h 10m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91bb840aca454ed4ac0339d6bbb678df : SUCCESS in 1h 45m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/df2786d33e894e1a82a263146aacd556 : SUCCESS in 1h 25m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/085ff50debb34862b143c7f85eda36ca : FAILURE in 1h 23m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e91597e3e09147d18aa63375272023aa : SUCCESS in 2h 36m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/626de5fcc0194ba587a8b27ea923295f : SUCCESS in 2h 27m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a0483a664124d158f974359be226b49 : SUCCESS in 1h 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6553a9de39ab48bab5c4a180e12a24fd : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e15073451184d34b815f650a37ad5ee : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edd3b1f22b5b4af6a7d57c2e4f944065 : SUCCESS in 2h 29m 14s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/19f4c14b3c79452c862756b5779a2f9f : TIMED_OUT in 5h 31m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d503527fa5158ea684c6cf00ad04c261677662c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 14:06:51.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"65dc77223e8e04a7d6fef3be9612406b61db02a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:15:05.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/988464 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"de4bb8922f8de464bab9ab4c4723c8273860b512","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 14:16:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce30fbf4569a57ad0c8b8315146d7a9a092b61ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 20:37:58.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9d2f38dba834465aee40c8b57ca068a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfa635d0a723411b9bca9e822f8a2cb0 : SUCCESS in 12m 14s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1749f82aabed4642a85fec7e0d8deb67 : SUCCESS in 1h 26m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/caa63c82cb734205b2ef0ed0098c72a6 : SUCCESS in 52m 47s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/fa13bfb654404e9cb0463fab71f96a91 : SUCCESS in 52m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f64260980d7c4c5589645b49630bd7ff : SUCCESS in 1h 02m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b5cd7a1c5f4427eb021f870b1a148af : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c59fcd422cc45bb95c7b552bc99acd1 : SUCCESS in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1259e3383b9479b9de3981578dc6ac4 : SUCCESS in 1h 43m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b6f66f2f9fd34447bcad3de72011a7c6 : SUCCESS in 1h 13m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d61c66a7913641b6b5e94c3c56486cd2 : SUCCESS in 2h 00m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc48f8b5896e4e1f9525cee3fefa3c63 : SUCCESS in 2h 31m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a3bf714bb154e07b66cb0483767333b : SUCCESS in 2h 18m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07c5068269264d00b406b68561ffe3d6 : SUCCESS in 2h 32m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd187d184554be3b497001c719eeed0 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c0d6af7eacb42e998ed16063db1449a : SUCCESS in 4m 18s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3b1271e03fe4b8e87c13e23f95bc727 : SUCCESS in 2h 37m 16s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d3896f2f7ac4076b925f7c27507c0a9 : TIMED_OUT in 6h 14m 27s","accounts_in_message":[],"_revision_number":5},{"id":"b77d8a59394d2f574d87120b0e1ef4aaaef99be2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 07:37:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"565aa55b59e4b4258e187e25cbcb975529f8cf8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 09:52:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0222dfab257548ae9ef2a211b4d61f87\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc11fc853698427bbf15355d3d4dfea3 : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cf0696f1b66d4a7e92cc771d9ad9818e : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/57752398e98443108244b39034f1935b : SUCCESS in 47m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0287342b9f064ca1b228bae00fa34ea2 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a88952c647b34c628fb64047a7bcc5ec : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/44285589c5624f798d485b1715f4c38b : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a46b22d7074f4d4ab29ab197d9ecb7ba : SUCCESS in 1h 39m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cac2c181dc63406780e6fd48c43faaea : SUCCESS in 1h 16m 34s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0916215b2c5444ca1be1eb104d2aefb : SUCCESS in 1h 10m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cdbf51516efa4ce693d1e32d5b5adb8a : SUCCESS in 1h 09m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1729da9691dd4af7a5ecd4ed7ec406b0 : SUCCESS in 1h 10m 24s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/743c2c2db1be4bc7b9cd0b63b3a74ab1 : SUCCESS in 1h 43m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4a219330f6341cea9266b045aadbd99 : SUCCESS in 2h 09m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a403ec4f34a3482fb32cbb61e924815a : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c371f659754c4b14b1c2c57176a58e70 : SUCCESS in 3m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b99bbd3b4ab4588a9da74f6b064f4c8 : SUCCESS in 1h 58m 05s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/bbdc18eac1504ae2ad3155138e891214 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9b7137d9c1c32acfbcf3afc8f28c91fba58ae5e9","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-05-14 11:41:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1b339d2e7baaa56c586bf7316de11a10560aeff5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-14 11:55:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9d424a0604342a017bc52e6657e8bdc585439a6f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 12:20:31.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4ce850b078f6d7b62754d3f541f17fc172defc23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:20:34.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88ff68983149470bbef2167a456f0e26\n\n\nWarning:\n  Change 988464,2 in project openstack/openstack-ansible does not share a change queue with 981149 in project openstack/openstack-ansible-os_skyline","accounts_in_message":[],"_revision_number":5},{"id":"90c3ea2a6274c71e9a8009d2e13761ac6210ab50","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 12:21:00.000000000","message":"Patch Set 5: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43259c95635bae526296c981795f4e833abb9a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:59:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9817c28d154487f8dcc587ecdf8594d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac495198467442d79068abed454c6edd : SUCCESS in 12m 01s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f8c44d01fe74ff59964fed1d3a38296 : SUCCESS in 1h 38m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b3a5885c417d468b9a675c76cc9447de : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9b714d6916f84dc4bb4daf0988cc8dd9 : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/81a101d9a6b44632b4645259ead641f8 : SUCCESS in 2h 01m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30fc88da824d4ccb97841865b01d1b2c : SUCCESS in 59m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbad17e332aa4dceb7783df72a729b09 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f26aefdb2b44f78eb5e6142f4db44f : SUCCESS in 54m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a80b23700c34b3fbf46c673dc771954 : SUCCESS in 2h 10m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e53034cdce344640bbe658686814e029 : SUCCESS in 1h 12m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73ce243e6dcb4cd8b81941ba4d52dfec : SUCCESS in 1h 19m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afe85bf6c12a446fb75883230ef2e51f : SUCCESS in 1h 11m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c8f9af81ae94dc89cf67e65a4d1ab69 : SUCCESS in 1h 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c875e043a484d5792ffc0c620554383 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63b80740a5ed48fbb547705d8a98e7b9 : SUCCESS in 4m 20s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e47e13e46df34596a8dbe2ee7dd966d5 : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/18642929987c49208c5e9616709cb89d : TIMED_OUT in 3h 14m 55s","accounts_in_message":[],"_revision_number":5},{"id":"f2767316eb5a090835dfdf831546ade1203433a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 18:19:39.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6bf23e0d1193d912c99ff9f74df65522a913f5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 21:06:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68e262b521924235ba11027ea3b32842\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fa1f49c018e495abf305df9034d1a1f : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5e5f05ba1dee42a79bbd27f0d494a8c0 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f38451acb81a464290c0587b3b691feb : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/89277984026942f2b6caf7b87b89e1ac : SUCCESS in 1h 13m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7c67a0fca9a542b8937ad1c1566e49ed : SUCCESS in 1h 00m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5689014e6f1549eea0242c357076767c : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7621c49538c43bca68c37e0dc203621 : SUCCESS in 50m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2d6d2df2e9d4f4ebc63b1995bffef72 : SUCCESS in 1h 44m 34s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fb919400c3d4b7cbc74df3670e3e7a7 : SUCCESS in 2h 01m 54s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc6affa2c5eb430c87167a09ee9dedc9 : SUCCESS in 2h 02m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1eb43fc4b2942b492f9f22fb8878804 : SUCCESS in 2h 22m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7000dd550c424221b1c7e93ff1b79cfc : SUCCESS in 1h 22m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e6c1561e7a242dda129bb56b4df8b6a : SUCCESS in 1h 53m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c3bcce2d804c96aea75040fed1e359 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d83f2dcab84c7697831c1b88d9b561 : SUCCESS in 4m 00s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ac5ea949a8746278fb7ac5b9f5b2783 : SUCCESS in 2h 37m 57s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/080afbd538254184be0350d758cffee2 : SUCCESS in 1h 44m 34s","accounts_in_message":[],"_revision_number":5},{"id":"6d0909ea051f4d0111238813b1e0291e9f66dcae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 21:09:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a57e9934afa9b578937a192cf922b2a7a8c3bcc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 22:49:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e1ef283dda348359f05bb8dd2d89e7a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d542061b7c543afb344f8dbab49a30e : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/05d339abfda14a47b933694ab066a3bf : SUCCESS in 55m 36s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8942a3d3fbbc4773834877aa2ea5f8db : SUCCESS in 55m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/26e55f7d34444206b5c26b2ec400d23b : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0e3a8da65de546189cf934046be15b0b : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/93279456d2374545a2ff7280558d67d9 : SUCCESS in 1h 38m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6793c4ac33844ce7a8d862e0298484c1 : SUCCESS in 1h 33m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31424116e1a7408ea86f4b05171ee967 : SUCCESS in 1h 29m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010d85051e064f34bcfe37b3521c2170 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/362d1ab1577c4f4f94e94ea4a0eacbc0 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"fbbf2df31be5840957e1b559f4fdb0a9ff3a9fea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 22:49:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0470c84c40755acdbbba64cdd00cc991d09a22a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 22:51:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eccbe94f98d94359b27062afde30f29e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6bbb349172e4dd49f97cd9e469532ef : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b759c9c8f3ed45462d9ff92cb41fb193babf9ffe","revisions":{"d55e12e5b0b84d969678292191b64e9490c1629b":{"kind":"REWORK","_number":1,"created":"2026-03-18 16:27:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/981149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/49/981149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/1"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"subject":"Fix skyline yarn-based installation","message":"Fix skyline yarn-based installation\n\nWith change of the logic for venv_build_host selection inside of the\npython_venv_build role, host expected to be builded wheels on and\nactually selected hosts were different, whic resulted in attempt to\nbuild wheels for skyline_console where yarn build was not run.\n\nThis resulted in wheels build failures and inability to install skyline\nwith yarn, as discussed in [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/972974\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/981148\nChange-Id: I387c99010a1daa58d3a968d74b2bf0956146ebc9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d55e12e5b0b84d969678292191b64e9490c1629b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d55e12e5b0b84d969678292191b64e9490c1629b"}]},"branch":"refs/heads/master"},"42029373d0717fd59d0e4525fee31030532ba8bb":{"kind":"REWORK","_number":2,"created":"2026-03-18 16:29:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/981149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/49/981149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/2"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:29:36.000000000","tz":60},"subject":"Fix skyline yarn-based installation","message":"Fix skyline yarn-based installation\n\nWith change of the logic for venv_build_host selection inside of the\npython_venv_build role, host expected to be builded wheels on and\nactually selected hosts were different, whic resulted in attempt to\nbuild wheels for skyline_console where yarn build was not run.\n\nThis resulted in wheels build failures and inability to install skyline\nwith yarn, as discussed in [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/972974\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/981148\nChange-Id: I387c99010a1daa58d3a968d74b2bf0956146ebc9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/42029373d0717fd59d0e4525fee31030532ba8bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/42029373d0717fd59d0e4525fee31030532ba8bb"}]},"branch":"refs/heads/master"},"e64b1f4a70530477a05977889cdf5515ddf37c02":{"kind":"REWORK","_number":3,"created":"2026-03-18 20:38:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/981149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/49/981149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/3"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 20:38:18.000000000","tz":0},"subject":"Fix skyline yarn-based installation","message":"Fix skyline yarn-based installation\n\nWith change of the logic for venv_build_host selection inside of the\npython_venv_build role, host expected to be builded wheels on and\nactually selected hosts were different, whic resulted in attempt to\nbuild wheels for skyline_console where yarn build was not run.\n\nThis resulted in wheels build failures and inability to install skyline\nwith yarn, as discussed in [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/972974\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/981148\nChange-Id: I387c99010a1daa58d3a968d74b2bf0956146ebc9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e64b1f4a70530477a05977889cdf5515ddf37c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e64b1f4a70530477a05977889cdf5515ddf37c02"}]},"branch":"refs/heads/master"},"76f1fc541759a61534afd8058b80f6c2fe42d47e":{"kind":"REWORK","_number":4,"created":"2026-05-12 12:19:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/981149/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/49/981149/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/4"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-12 12:19:39.000000000","tz":120},"subject":"Fix skyline yarn-based installation","message":"Fix skyline yarn-based installation\n\nWith change of the logic for venv_build_host selection inside of the\npython_venv_build role, host expected to be builded wheels on and\nactually selected hosts were different, whic resulted in attempt to\nbuild wheels for skyline_console where yarn build was not run.\n\nThis resulted in wheels build failures and inability to install skyline\nwith yarn, as discussed in [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/972974\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/981148\nChange-Id: I387c99010a1daa58d3a968d74b2bf0956146ebc9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/76f1fc541759a61534afd8058b80f6c2fe42d47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/76f1fc541759a61534afd8058b80f6c2fe42d47e"}]},"branch":"refs/heads/master"},"b759c9c8f3ed45462d9ff92cb41fb193babf9ffe":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-05-13 14:06:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/981149/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/49/981149/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/49/981149/5"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-18 16:26:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-13 14:06:51.000000000","tz":0},"subject":"Fix skyline yarn-based installation","message":"Fix skyline yarn-based installation\n\nWith change of the logic for venv_build_host selection inside of the\npython_venv_build role, host expected to be builded wheels on and\nactually selected hosts were different, whic resulted in attempt to\nbuild wheels for skyline_console where yarn build was not run.\n\nThis resulted in wheels build failures and inability to install skyline\nwith yarn, as discussed in [1]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/972974\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/988464\nChange-Id: I387c99010a1daa58d3a968d74b2bf0956146ebc9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/b759c9c8f3ed45462d9ff92cb41fb193babf9ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/b759c9c8f3ed45462d9ff92cb41fb193babf9ffe"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
