)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~981838","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~stable%2F2025.1~I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a","project":"openstack/openstack-ansible-os_skyline","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-20 20:16:34.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-20 20:16:34.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-13 10:49:55.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-20 14:39:06.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-20 10:38:58.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a","subject":"Disable upgrade jobs from 2024.1","status":"MERGED","created":"2026-03-23 21:32:46.000000000","updated":"2026-04-20 20:18:36.000000000","submitted":"2026-04-20 20:16:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981838","meta_rev_id":"61089199ab906a827d2abd495c8357a4f3a44127","_number":981838,"virtual_id_number":981838,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 20:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-13 10:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-20 10:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-20 10:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 21:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 10:07:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-13 11:01:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-04-13 11:01:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-04-13 11:01:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"52ad4b908fa6c8c08df049a068d97725a16a2f9a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 21:32:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82417f5a56abadc106d2e1608159e6cebdca9c1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 21:33:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2d812286f24f3021fec77da8c4cd3bf7c8814583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 21:34:16.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/openstack-ansible-os_skyline on branch stable/2025.1.  The\nproblem was:\n\n  The key \"check\" appears more than once; duplicate keys are not permitted.\n\n  templates:\n        - check-requirements\n        - openstack-ansible-linters-jobs\n        - openstack-ansible-deploy-aio_metal-jobs\n        - publish-openstack-docs-pti\n        - build-release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy\n  ...\n\n  in \"openstack/openstack-ansible-os_skyline/zuul.d/project.yaml@stable/2025.1\", line 17","accounts_in_message":[],"_revision_number":2},{"id":"4c430dce209f8e1a4e76e19037411a020d59a704","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 21:34:54.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c12e5636a5b8fc353f9162abfc9e56cccbc46f22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 00:20:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6d1510503c464ba68e256c67fb412a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fc762819504467d9e4c2dd6b107e569 : SUCCESS in 8m 01s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f83ce21985f84e17a629d1fafea8f391 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bcd5d878b8c2439a8bda85ca1645e2c1 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ac38fa2f4334a03b23f5f9996e12b03 : SUCCESS in 1h 32m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f8c27342cad46d8a121ce8995121a80 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c67ec6cd4ec48f08681b11d55063866 : SUCCESS in 1h 17m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b569cba74fc491dbc76fdf061797abf : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad635a3888164ba5ba73728382e81cbd : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16d43ec6243e4bd0bf7cb6c49444786f : SUCCESS in 46m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7dcf83ebe26d48e69a25d53d4682783b : SUCCESS in 1h 40m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd16bc97c4746ef894cb0fd606f89b4 : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a67a73b4cd8242ba99c1dec15d873dde : SUCCESS in 47m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6132beef532045228e74b9fe3ea489a3 : SUCCESS in 1h 00m 39s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/289442b014154a539a49f5162409977f : FAILURE in 39m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89f6ab2ea339412699f3a36f55e3abf1 : SUCCESS in 1h 59m 54s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7739c1568cea4914a96b2667ea867f57 : SUCCESS in 2h 09m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca3ca31ff2704c409a6855d520820fc4 : SUCCESS in 2h 24m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/754fa525b0294534beadd7ee95fd4ab4 : SUCCESS in 2h 05m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886c1aabad18474d9cb09f0ad7bfe1f8 : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/318082ea8fd74b92b67ee3ed7069d506 : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ad4a5852d77489998192a20d9584c76 : SUCCESS in 2h 34m 24s","accounts_in_message":[],"_revision_number":3},{"id":"97e68f2c0bf485633a3280031d2d8f6bc7bb4bc9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-10 19:39:51.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"971c2dfc65d485280a60167d0f121b233b4f6434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-10 19:40:12.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a9162d9e7f40baf8e727d37b3c892d82026b7f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:27:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419593e8521d4768b126627a3745e98f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0828a2c1a9eb46d9bdca4ede212305d2 : SUCCESS in 12m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9c601ac1bd234d47b7a01972dbb72fb2 : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7d54ae87f3fa4fb7ae2f84349197caf9 : SUCCESS in 45m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/280e04d7dfc146e4890891148b6acd96 : SUCCESS in 1h 28m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/93cd40a5d50c4852b3a0fb0a19f14762 : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e91acb484f8540c0b0f7dbc1376fff77 : SUCCESS in 44m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cb85d49e65124ccaa60ff6807dc6ca96 : SUCCESS in 45m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/706e61cc906748e2a8ad7178695b88d0 : SUCCESS in 41m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f07ec421f9d84289a45b51dffe3af87e : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f1d24d7ce4074702831ce836d5b5494d : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c80e2a2a418441d7a64e809d885f3bf9 : SUCCESS in 1h 17m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98e57d7ff6054a01af1aac89cb527479 : SUCCESS in 1h 16m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc828b42325c4ac6a2ab28ab4aa6896e : SUCCESS in 1h 01m 31s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d991ba6fc0ba4d969ac7517108db5152 : FAILURE in 1h 14m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80edb2f3dfb64ba9af50cc15901ed659 : SUCCESS in 59m 16s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60e69f0516a24195aa1efbe9b6be3baa : SUCCESS in 2h 31m 44s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c9a46cb59634125baa4c4dea8cea9fe : SUCCESS in 2h 32m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b87820544954fbc8c850310fce297b6 : SUCCESS in 1h 55m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9380296d034af68ac972a81f124445 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f043f438c2e54f1e9e812a501ca39d22 : SUCCESS in 1h 49m 43s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9359c40a5d1f47db900ebb25e639c97b : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":4},{"id":"969aef5a42ebe18ca95c0aba5ec58ceb0fd5c38c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-13 10:49:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3bf282cde5b1aa36dff989dc4743f422b04958ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-20 10:38:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"da8def008a49809646ac24774f5fe97f01342467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:42:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a35eb5228b4289cd9007fcd6fb137c64a10f8096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 14:38:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/978c7b7df8344366915396555c1357db\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c69454658f804974a6ff4817e3eeea8a : SUCCESS in 14m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/16eb740255474a29b9dddce556a949b6 : SUCCESS in 1h 31m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e436a3f407e84a2781a1a927c7a05fcf : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21efa5b966ae4239942be5fd67d92a3e : SUCCESS in 52m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0664c1186c0402d80ba91f8ad40936e : SUCCESS in 53m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b2a1d2b0dcd744eca75a30706a5eb68d : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a40f5ce7a3437ab4707099b76fff71 : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ab549746d4b4bda9c45d311a33043e9 : SUCCESS in 1h 28m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cdd2718db03451f8d054583a934c6a4 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbcc87844dc24aab9562806e43a7acf3 : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34a742ad7b6044dcb3a36c7fadc89ffe : SUCCESS in 1h 34m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ed59a124bc4ea1a55048b92c8e9adb : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4a1ed9737a641cb8a3f1d1b2202cc2c : TIMED_OUT in 3h 11m 56s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eaca75bb754f44aea6857da0f1ec266f : SUCCESS in 2h 45m 41s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4e0aa5263e447c685cc868e111f4503 : FAILURE in 1h 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe809b7363bac7fc8b1c12e56f0cc4798271e794","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-20 14:39:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c4cc07932f8b74e4fe087507bdc5dfb089e0d5a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:38:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b69eb6cd82874f2886e43c4b05b58cd7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49559a7b3d3f4bcf84e99532748b342a : SUCCESS in 12m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d3007c9ba72402a807cc01d9953b298 : SUCCESS in 51m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e53b3cffd1ee47eead4e7d542d61add2 : SUCCESS in 47m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9d5bfd6694646d7a51780d1a443b966 : SUCCESS in 1h 34m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0e4b858b4b6418698672029c60c37b3 : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f2d2858c94f48568f44ecaf95ce4291 : SUCCESS in 53m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3c6983548754f5c8c08e33cee388623 : SUCCESS in 52m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ccdae3dc614434f903abcda66c85d2b : SUCCESS in 1h 26m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e49bde5decf432d91e165dcdb452259 : SUCCESS in 1h 43m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8421a71e4104165bb3f7a4ac811c562 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb1dee0aab547b1bcddfb1be925c3e0 : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35773cb39ea246fa91e7d67f08f15a65 : SUCCESS in 1h 42m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b03548dd9df1438f93c2c2ba2d202cdd : SUCCESS in 2h 22m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c16f6045a8304e4ca4c426edf92dde78 : FAILURE in 1h 00m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a336d85bea84eebb897e5b8e5ef5220 : SUCCESS in 1h 54m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66417b39b00a47a0acca706b3e4bbef3 : SUCCESS in 1h 14m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57b06ae535b44c2a82862f6a4cbd58cf : SUCCESS in 1h 10m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f69e9a201fd643d587d3be7846fd1e28 : SUCCESS in 2h 15m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b473e5058b406390f3b1b892436975 : SUCCESS in 8m 11s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d26c541803f40c7ad99ef52fbad9144 : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af6f3d51697b4b069b5549693c833ea8 : SUCCESS in 2h 28m 07s","accounts_in_message":[],"_revision_number":4},{"id":"d8d3824de6c540203b311cbb1a50c7e7aa2526b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:41:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"adbf7073ebcecb466d9350e99efc3c5c94c6a1d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:16:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/040803154fb140229bd525bcfc23513e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba7f69cf6da34dabb3f7b1aea9727833 : SUCCESS in 13m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/22e623c8e79e41d2a6af79090630cc3c : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1a6f86b38e484a21a521d0ac6bee24f5 : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ace0d82093334d869edb5bf4f4bf8b50 : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eaeba44a54b24e2e8a5cb701e3f03031 : SUCCESS in 1h 13m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3100cd2dd5b486b813952bf0f116c1d : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4fd509617bb4c2e99b68e5a63c668af : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/069ea9a99e334b1f82de0bf699a90ac8 : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f95c0ad226694ff6930cb7830a545ab5 : SUCCESS in 1h 34m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fedfb5563c749b689ad95cf146b3e95 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/961881560c50496e91040e7e06fe4319 : SUCCESS in 1h 32m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7b3c1b680b414ca4d92f8dd88efebe : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbf4dca7791b4d5abaffdb5f93dae878 : SUCCESS in 2h 18m 27s\n- openstack-ansible-deploy-aio_skyline_yarn-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5917e35ab17478c84e9d64467d8ad0d : SUCCESS in 2h 30m 08s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7dc96491de0a4aad9a006dc5a1e9ec03 : FAILURE in 1h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"71e9dafec62431279fbff1b3a28299a07c3c6df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:16:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"61089199ab906a827d2abd495c8357a4f3a44127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:18:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a4c43b6b4b453ab64565dc724e3552\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d89a86a8e4e4e07af24660087606c7e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d4a935e7d31e36d87b750c60c7d284837f37be7f","revisions":{"f05721b717e048845caa18caa6a24b3e6faf773a":{"kind":"REWORK","_number":1,"created":"2026-03-23 21:32:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/981838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/38/981838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/1"}}},"commit":{"parents":[{"commit":"d9094e3ecab73db3248cdf8e832698ad9782c6ce","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d9094e3ecab73db3248cdf8e832698ad9782c6ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:32:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:32:46.000000000","tz":0},"subject":"Disable upgrade jobs","message":"Disable upgrade jobs\n\nChange-Id: I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f05721b717e048845caa18caa6a24b3e6faf773a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f05721b717e048845caa18caa6a24b3e6faf773a"}]},"branch":"refs/heads/stable/2025.1"},"45c391ec8d35fd8a691be272709a2ffe6c5e43f5":{"kind":"REWORK","_number":2,"created":"2026-03-23 21:33:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/981838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/38/981838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/2"}}},"commit":{"parents":[{"commit":"d9094e3ecab73db3248cdf8e832698ad9782c6ce","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d9094e3ecab73db3248cdf8e832698ad9782c6ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:32:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:33:41.000000000","tz":0},"subject":"Disable upgrade jobs","message":"Disable upgrade jobs\n\nChange-Id: I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/45c391ec8d35fd8a691be272709a2ffe6c5e43f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/45c391ec8d35fd8a691be272709a2ffe6c5e43f5"}]},"branch":"refs/heads/stable/2025.1"},"974964c4d4aa859ba361ab125eb2e23c519f9cd5":{"kind":"REWORK","_number":3,"created":"2026-03-23 21:34:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/981838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/38/981838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/3"}}},"commit":{"parents":[{"commit":"d9094e3ecab73db3248cdf8e832698ad9782c6ce","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d9094e3ecab73db3248cdf8e832698ad9782c6ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:32:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:34:54.000000000","tz":0},"subject":"Disable upgrade jobs","message":"Disable upgrade jobs\n\nChange-Id: I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/974964c4d4aa859ba361ab125eb2e23c519f9cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/974964c4d4aa859ba361ab125eb2e23c519f9cd5"}]},"branch":"refs/heads/stable/2025.1"},"d4a935e7d31e36d87b750c60c7d284837f37be7f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-04-10 19:40:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/981838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/38/981838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/38/981838/4"}}},"commit":{"parents":[{"commit":"d9094e3ecab73db3248cdf8e832698ad9782c6ce","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d9094e3ecab73db3248cdf8e832698ad9782c6ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-23 21:32:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-10 19:40:12.000000000","tz":0},"subject":"Disable upgrade jobs from 2024.1","message":"Disable upgrade jobs from 2024.1\n\nChange-Id: I5e3bcc8e63c19732ae1252e10e92a36d0ec0277a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4a935e7d31e36d87b750c60c7d284837f37be7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4a935e7d31e36d87b750c60c7d284837f37be7f"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
