)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~982629","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~I3196e79974146db95ab4ae93cba6cf2d6a31e930","project":"openstack/openstack-ansible-os_skyline","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-05-18 11:49:26.000000000","reason":"removed on reply"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-05-19 09:20:35.000000000","reason":"Change was submitted"},"39035":{"account":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"last_update":"2026-05-19 09:20:35.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-19 07:31:17.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-19 09:20:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3196e79974146db95ab4ae93cba6cf2d6a31e930","subject":"Fix Object Storage section for RGW","status":"MERGED","created":"2026-03-30 10:47:04.000000000","updated":"2026-05-19 09:21:46.000000000","submitted":"2026-05-19 09:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982629","meta_rev_id":"13cfeed09b93851f3893e33e5e2664acc77a45c1","_number":982629,"virtual_id_number":982629,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 09:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"all":[{"value":1,"date":"2026-05-18 11:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-05-13 14:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":2,"date":"2026-05-19 07:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-18 11:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":1,"date":"2026-05-19 07:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},{"value":1,"date":"2026-05-14 10:27:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 13:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 14:35:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-30 14:35:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-05-12 10:27:58.000000000","updated_by":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"reviewer":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"state":"REVIEWER"},{"updated":"2026-05-18 11:48:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-18 11:48:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-05-18 11:48:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"adf688c2af362982fef0ca41996802232450e519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 10:47:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49e4fdd1a7f6e174df48d6d392ee0cd10e0a565d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 11:16:47.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"5df62d892076e0fe050f3bf3d563e7ee9444a260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 13:49:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2e07f8088f7434491ce968558109b02\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/861bcd9579604bf7b9bf46e91e62b8e5 : SUCCESS in 9m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3919a3e7edc84786b205698cccb2fd96 : FAILURE in 48m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/31b8b7390aed4e8b82d1a88bc0ba5a3c : FAILURE in 1h 50m 25s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f33a427c20c3406da748ba323a7bf784 : SUCCESS in 1h 46m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/916a4eb816ad4187bb8205427e09d06f : FAILURE in 1h 10m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c1b8f8ac06054a76a660dd7c27e4582e : FAILURE in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3de279e4b2fd40febeda5330cf50876d : SUCCESS in 1h 51m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b414279787f44111b65ff997995c300d : SUCCESS in 1h 47m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d9eb0a1ae894dfbb1b3f1868046c722 : FAILURE in 1h 07m 26s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8bdeadaace7e4bc8b589c228466d5c53 : FAILURE in 1h 42m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fc556ca692f41479ac2c58e14087e0c : SUCCESS in 2h 29m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4047167920043dba8ccd0b52c542af2 : SUCCESS in 2h 36m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75fddeee671342729f3b39b161e0d8cb : SUCCESS in 2h 39m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248313f4fae54385954881c21e14c3ae : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a29f17f9b5a04f0aa5497b2e7a9c6fe5 : SUCCESS in 2h 55m 25s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/3aed7a151be54f7d96c47e2fba1233e8 : SUCCESS in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3625757dd047b8c2d2982b1bdfad89906fb43520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 09:03:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d89fd0ddb7d989d98c29da314966610b6eb67dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"date":"2026-05-12 10:27:58.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"25535e7dfd8c2f5204c797e63636af3e436b62f7","author":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"date":"2026-05-12 10:30:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89297122b551df662b57a10b75b877a57e2226c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 10:35:39.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6fbab8329d6e3d50e5c88ae095ee15bf382d0878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:40:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4754cd3939584e2ea3e532526e0617b1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b61279b8ce4ef3a2af2cb26e8392ed : SUCCESS in 9m 19s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7c89545087dd4b1ca8983841237b8d04 : FAILURE in 1h 13m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d4843681e1242f2b954826488fc8149 : FAILURE in 1h 28m 16s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8e709d0050d84513a0514f649f04c216 : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4dbd7eff7370414391eb832e418d1eae : SUCCESS in 1h 20m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5d272b93ba7c40d8a5cbc39edea5c15e : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4178627cce2747c9836f2030085213f3 : SUCCESS in 55m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3015ccf53c04a31993a9fec21539e17 : FAILURE in 1h 31m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9364bbf14bf4d26813f0360c9abc2c4 : FAILURE in 2h 10m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ea9befb117ee45ed9d942bc8e6150b67 : FAILURE in 2h 22m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0fba3cf7a7b431b904d8c265c37446f : SUCCESS in 1h 16m 37s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/211bfa119a4f489daa4f02662e666ca1 : SUCCESS in 1h 16m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e2498aa037c4e699bdb2a8d3b944109 : SUCCESS in 1h 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4132923fd45a468f93dba74464f68fd7 : SUCCESS in 6m 32s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14a97b2a309b4de3b46b68460558e2e2 : FAILURE in 2h 19m 41s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/addace947a5b4a86843d317ba3cde9cd : FAILURE in 3h 02m 56s","accounts_in_message":[],"_revision_number":3},{"id":"9da13db29165d1f2ca0a7eff85c1658422114326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 14:38:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_39035\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"}],"_revision_number":4},{"id":"dcf2932e543d7846356c0deaf39393b93572de00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:57:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c0155f4067a43849ffc32cf5dfb41e1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ed384ec7c7743a89a908250ddd96b35 : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/67ff9b99f3d948cfb80f63035cace60f : SUCCESS in 1h 45m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cf86be1a35974e479b3fbff433ef06c6 : SUCCESS in 1h 48m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2e3b733bd9fe452f95d81ac475b89ae7 : FAILURE in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eafb1fbcbf3c48dcad2525367b0c7081 : FAILURE in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30c0e1199a3c493f82cc92bb47d796f2 : SUCCESS in 59m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/325f31adccb14e62b5f601e3e014b037 : SUCCESS in 1h 51m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/709718a194ea4fc3a330b150195d719f : SUCCESS in 2h 31m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5886e8d5b3c445a8ac16d46e8225ca1c : FAILURE in 1h 54m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e433749b8734a9c92c5cde4228f6c52 : FAILURE in 1h 55m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3226738e3c54239a1d33be6f9b72522 : SUCCESS in 2h 32m 23s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8172c5b1549b42d7bc2fd461101b1414 : SUCCESS in 2h 48m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62427169a0a041e19b440c6e4bc33132 : SUCCESS in 2h 53m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482530df1e534860aa918f6905e3d40c : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f65e84d210c48829094b998696f8cbb : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/5ec3ead8a6e24c319bd39c176cb67041 : TIMED_OUT in 5h 13m 12s","accounts_in_message":[],"_revision_number":4},{"id":"7f5b1c7ca9612c40d390bd567ff5282e657aafed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 14:09:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_39035\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"}],"_revision_number":5},{"id":"690a42a4ed7ee37c4d9ccb4fdcbe11022bf47a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:15:07.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-os_skyline/+/981149 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"b2502ce6ab0f0fe73572368cbf361b97ccecd21b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 10:27:55.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"891712347241ae24f203c25562beff627b5978f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 13:16:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7839adb134743209b36ab494f32e80d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5733dd6b41c4a3cbf90bfaf2f28b92d : SUCCESS in 11m 19s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3bc94f5091794d77af7b8a8ac32d1dc3 : SUCCESS in 1h 15m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9f11c3bb9ef34b0ab1b7d87f65d83524 : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/fe6e2f7f414046479d7876d85fd3986c : SUCCESS in 1h 31m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed45e776743b4c4a9ac037b7bd5ee8c8 : SUCCESS in 49m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/175e009149a34e3aa95e4c76896bfa65 : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a5ea4332d28491c9c35607524e72d43 : SUCCESS in 1h 39m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab853b1699514cf495065133bdef0009 : SUCCESS in 1h 34m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42ecb8b6b3f144ccacb10b474c5e99a7 : SUCCESS in 2h 01m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/583fb55f21b04271b49ef100a02acfc4 : SUCCESS in 1h 07m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d3ad6deadf41f4a531e44538b4127a : SUCCESS in 2h 28m 48s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cdc954a1dcf46bd88c88fa650d62ce3 : SUCCESS in 2h 40m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaf620d638b2446b928575ff036a1a6c : SUCCESS in 2h 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54cf5128f8cb40a6b547a1521940ee41 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27c9c20183d34d9f885b5621afbd1985 : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f14967e526e453e923391abe45504c9 : SUCCESS in 1h 39m 34s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/009a243a48b44602b905cac901f18257 : SUCCESS in 2h 27m 26s","accounts_in_message":[],"_revision_number":5},{"id":"bf13690db71d596c582eb18bc683ee3bf029af3c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-18 11:27:12.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b80b802b39edb409383ff0249a7bd3709f5182a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-18 11:29:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ffa4d37f07dac4c1b3b582c569c1ad4410894511","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 11:48:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8bfb73db64862b8641a63fd58ff4e411f1031d9a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-05-18 11:49:26.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9431db28ef8a4b42a51c947ae0b1c6002dc23175","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-19 07:31:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5e3c23d68094a7f9b3bcf476b696bfabe5429998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:34:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d819530a0eaf29b84121bb853e7dabb5abf55247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d764ff2724a44a3a195a21badd0b9e0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a3d42f15b8a4e609e73136a11f43498 : SUCCESS in 11m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/13fa9ecb743543edad8cf89bc19cfb19 : SUCCESS in 54m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a42e846d44e9440ebcea3b2704dd4eef : SUCCESS in 1h 28m 59s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/62d03f4c32134bc0ab3bcd2702851c2b : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f84428136ef84a34984dfe2f239f1a75 : SUCCESS in 53m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fccd10a210ef456cb5c62e5f0702cb28 : SUCCESS in 54m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba1123c2cbab4fefa3cab475b8bd71e8 : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c893743488a48da805fd7f1ebf4293c : SUCCESS in 1h 35m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb934e66f344ffd960730933855e505 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/703ff1107a6e4b2496994e4df85d9213 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":5},{"id":"473006426792fd9edbc443ef31699bf2949eb98a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"13cfeed09b93851f3893e33e5e2664acc77a45c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:21:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03bf8a99320c475a86238408a60bc9a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f522cc019e5f4059b5c774b2a6b03a9f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f2c95a07e13c4f001edd6f3f5270df1f36885b94","revisions":{"37e84b9eee9ee45f7988451e563d8d87b899b5dd":{"kind":"REWORK","_number":1,"created":"2026-03-30 10:47:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/982629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/29/982629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/1"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"subject":"Fix Object Storage section for RGW","message":"Fix Object Storage section for RGW\n\nWhen RGW is deployed with the Swift API,\nthe URI to which redirect should be happening in /swift\n\nWith that, current logic of _skyline_service_locations is not taking\nendpoint path into the account, as is might be polluted with irrelevant\nto rewrites parameters.\n\nThus, we create a special case for Swift, when URI should be appended\nwith /swift, and be just / for all other scenarios.\n\nChange-Id: I3196e79974146db95ab4ae93cba6cf2d6a31e930\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/37e84b9eee9ee45f7988451e563d8d87b899b5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/37e84b9eee9ee45f7988451e563d8d87b899b5dd"}]},"branch":"refs/heads/master"},"463ea41a3da86558b4c72fa1b3d0cbabaab9b07c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-12 09:03:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/982629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/29/982629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/2"}}},"commit":{"parents":[{"commit":"e64b1f4a70530477a05977889cdf5515ddf37c02","subject":"Fix skyline yarn-based installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e64b1f4a70530477a05977889cdf5515ddf37c02"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-12 09:03:24.000000000","tz":0},"subject":"Fix Object Storage section for RGW","message":"Fix Object Storage section for RGW\n\nWhen RGW is deployed with the Swift API,\nthe URI to which redirect should be happening in /swift\n\nWith that, current logic of _skyline_service_locations is not taking\nendpoint path into the account, as is might be polluted with irrelevant\nto rewrites parameters.\n\nThus, we create a special case for Swift, when URI should be appended\nwith /swift, and be just / for all other scenarios.\n\nChange-Id: I3196e79974146db95ab4ae93cba6cf2d6a31e930\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/463ea41a3da86558b4c72fa1b3d0cbabaab9b07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/463ea41a3da86558b4c72fa1b3d0cbabaab9b07c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"37e84b9eee9ee45f7988451e563d8d87b899b5dd","theirs":"e64b1f4a70530477a05977889cdf5515ddf37c02","contains_conflicts":false}},"7094788e027da04add98d53cfa03e4f47546e62d":{"kind":"REWORK","_number":3,"created":"2026-05-12 10:27:58.000000000","uploader":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"ref":"refs/changes/29/982629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/29/982629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/3"}}},"commit":{"parents":[{"commit":"e64b1f4a70530477a05977889cdf5515ddf37c02","subject":"Fix skyline yarn-based installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/e64b1f4a70530477a05977889cdf5515ddf37c02"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"committer":{"name":"velychkovsky","email":"k.velychkovskyi@cloudfarm.tech","date":"2026-05-12 10:27:35.000000000","tz":120},"subject":"Fix Object Storage section for RGW","message":"Fix Object Storage section for RGW\n\nWhen RGW is deployed with the Swift API,\nthe URI to which redirect should be happening in /swift\n\nWith that, current logic of _skyline_service_locations is not taking\nendpoint path into the account, as is might be polluted with irrelevant\nto rewrites parameters.\n\nThus, we create a special case for Swift, when URI should be appended\nwith /swift, and be just / for all other scenarios.\n\nChange-Id: I3196e79974146db95ab4ae93cba6cf2d6a31e930\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nSigned-off-by: velychkovsky \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/7094788e027da04add98d53cfa03e4f47546e62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/7094788e027da04add98d53cfa03e4f47546e62d"}]},"branch":"refs/heads/master"},"d8be88450811fd49ba42f2b5d76134b1e921ffd3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-12 14:38:48.000000000","uploader":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/982629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/29/982629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/4"}}},"commit":{"parents":[{"commit":"76f1fc541759a61534afd8058b80f6c2fe42d47e","subject":"Fix skyline yarn-based installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/76f1fc541759a61534afd8058b80f6c2fe42d47e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"committer":{"name":"Kostiantyn Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","date":"2026-05-12 14:38:48.000000000","tz":0},"subject":"Fix Object Storage section for RGW","message":"Fix Object Storage section for RGW\n\nWhen RGW is deployed with the Swift API,\nthe URI to which redirect should be happening in /swift\n\nWith that, current logic of _skyline_service_locations is not taking\nendpoint path into the account, as is might be polluted with irrelevant\nto rewrites parameters.\n\nThus, we create a special case for Swift, when URI should be appended\nwith /swift, and be just / for all other scenarios.\n\nChange-Id: I3196e79974146db95ab4ae93cba6cf2d6a31e930\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nSigned-off-by: velychkovsky \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d8be88450811fd49ba42f2b5d76134b1e921ffd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d8be88450811fd49ba42f2b5d76134b1e921ffd3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"7094788e027da04add98d53cfa03e4f47546e62d","theirs":"76f1fc541759a61534afd8058b80f6c2fe42d47e","contains_conflicts":false}},"f2c95a07e13c4f001edd6f3f5270df1f36885b94":{"kind":"NO_CHANGE","_number":5,"created":"2026-05-13 14:09:30.000000000","uploader":{"_account_id":39035,"name":"Kostiantyn Velychkovskyi","display_name":"K Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","username":"velychkovsky","status":"CloudFarm Co-Founder / CTO"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/982629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/29/982629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/29/982629/5"}}},"commit":{"parents":[{"commit":"b759c9c8f3ed45462d9ff92cb41fb193babf9ffe","subject":"Fix skyline yarn-based installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/b759c9c8f3ed45462d9ff92cb41fb193babf9ffe"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 10:46:58.000000000","tz":120},"committer":{"name":"Kostiantyn Velychkovskyi","email":"k.velychkovskyi@cloudfarm.tech","date":"2026-05-13 14:09:30.000000000","tz":0},"subject":"Fix Object Storage section for RGW","message":"Fix Object Storage section for RGW\n\nWhen RGW is deployed with the Swift API,\nthe URI to which redirect should be happening in /swift\n\nWith that, current logic of _skyline_service_locations is not taking\nendpoint path into the account, as is might be polluted with irrelevant\nto rewrites parameters.\n\nThus, we create a special case for Swift, when URI should be appended\nwith /swift, and be just / for all other scenarios.\n\nChange-Id: I3196e79974146db95ab4ae93cba6cf2d6a31e930\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nSigned-off-by: velychkovsky \u003ck.velychkovskyi@cloudfarm.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f2c95a07e13c4f001edd6f3f5270df1f36885b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f2c95a07e13c4f001edd6f3f5270df1f36885b94"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d8be88450811fd49ba42f2b5d76134b1e921ffd3","theirs":"b759c9c8f3ed45462d9ff92cb41fb193babf9ffe","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
