)]}'
{"id":"openstack%2Fopenstack-ansible-os_skyline~985595","triplet_id":"openstack%2Fopenstack-ansible-os_skyline~master~Ibf41cf2d902b239189f8647c9e67f654f6238d38","project":"openstack/openstack-ansible-os_skyline","branch":"master","topic":"openstack/openstack-ansible-sync/sync-files","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-24 22:50:34.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ibf41cf2d902b239189f8647c9e67f654f6238d38","subject":"Updated from OpenStack Ansible Integrated repository","status":"MERGED","created":"2026-04-20 23:16:02.000000000","updated":"2026-05-25 13:56:23.000000000","submitted":"2026-05-25 13:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985595-openstack/openstack-ansible-sync/sync-files","meta_rev_id":"5201966cd35d11637bc059d297908eeafd42e573","_number":985595,"virtual_id_number":985595,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 13:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-25 09:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-25 09:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 03:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 21:38:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-25 09:39:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-05-25 09:58:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"d5e01a1d0ef3d0ff830399be15e7eacf4dddef25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 23:16:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0890ed819278be0fbcc81896f0cc32abb4e105cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:43:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56cc241f05934341b02433f89e506386\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec62e748c9b645baa751bc974c2e506a : SUCCESS in 12m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b17a19502fc0413c85e491b3b07aecb2 : FAILURE in 1h 05m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0996555b17c04faca0ab01352f78d695 : SUCCESS in 1h 35m 18s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/dee2567007ce4fcfb194107139bcd630 : SUCCESS in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/685dce66ba9a4447bc9a48aead01b697 : FAILURE in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae5aa44374934e4a9ddd3cf0a1aadeff : FAILURE in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2a062c913484d70a6ae10376e8c4f70 : FAILURE in 1h 33m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbccce15d83c405ca94e8a6c0e009293 : SUCCESS in 1h 44m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cff376c1bef84277a3ca58312eb98cae : FAILURE in 2h 05m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f10d5d33cd0342409439447fac9e04d2 : FAILURE in 1h 14m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b85718a17154c74ba0ad26c6dc113c9 : SUCCESS in 2h 26m 43s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff5d594da1c44fc7a1951e2b963723b4 : SUCCESS in 1h 15m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e61f16ccd7ed4b56b00ffd3d59cc5fb3 : SUCCESS in 1h 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96aee4ec3a8142bfb77daebd9064a6fc : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccbd6186e9664401aa45d7de5414fe52 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a33455fc05c44738b7a8511a54818d9 : TIMED_OUT in 3h 11m 40s (non-voting)\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/b9b139ae425041cca0605dfd7b6cf76d : SUCCESS in 1h 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aeb7160ca5363708f78d8157ac5900d74039f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-19 21:38:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"9a2bccd52b7b9923928b2509245fd551369f4e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 01:00:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1ea147ccbe749d998aa2f2067f3b44a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b11ab7b6495046a6812ecb7c37d94873 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1e3e8e0bb2254cd79d66b18d74037e10 : SUCCESS in 51m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/88e3baabbd944ce297a7f25368a318f9 : SUCCESS in 1h 38m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/19acbe62a35845a3a85ee3ed3dae84d9 : SUCCESS in 1h 36m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/75a038638c6c400883139e18378db4b2 : SUCCESS in 51m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a361f04cefaa44518f035d56c4ca4293 : SUCCESS in 1h 39m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f09b0f89e044415899b3abe8f0f35ac : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/595b402acc754f6eab2243ece25513f0 : SUCCESS in 1h 40m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e299c8aae554b8dadb9c494d9da8161 : SUCCESS in 1h 48m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c82d4367d17d42d7b3c90cc599b91d83 : FAILURE in 1h 46m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bde0aaa43e414eb093a3f4853f6b105e : SUCCESS in 1h 11m 37s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fae4b04c6e1949b1b35c552d0f6a385a : SUCCESS in 1h 52m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baf15582ba274f2ca8daeb636cb34df6 : SUCCESS in 2h 30m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc0c11c52204c0c8f3ef0f59894ee4a : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a030b597608419bb337a44bc07a1b05 : SUCCESS in 3m 10s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90dc9d4e1585431f81afa3b7dd4129b9 : SUCCESS in 2h 40m 20s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/10fa116d22164909bd67c53c399063ac : TIMED_OUT in 3h 18m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8139835ce0827df622168cd495179875c0652033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-22 10:02:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":3},{"id":"bd1b78871f1d8c2dc78c54c24c2ef91702e84079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:24:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40e57ee633d3408da7f018671deb5fcf\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68a0d75a0ac84d408a1155c3a52a0562 : SUCCESS in 8m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ed42d28ab8834442a7ab40c7584d8369 : SUCCESS in 1h 27m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e5dddfd6fd1499d8acf1bc2af171858 : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0a642706395d4c50b6f6fcc2b900f8fc : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c73ae41a89094da7a4d23c1d820afbfb : SUCCESS in 55m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b9cfd6ff8364e33b863e386c216a7fb : SUCCESS in 53m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7f4e49561524b7cbda52b4276427549 : SUCCESS in 50m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6f7810d7afb45628645c6a7b8fa7048 : SUCCESS in 1h 22m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ce525811f7c843778606aba6df442f58 : SUCCESS in 2h 24m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa49d99a8cc24bdfa66ebb2c8a8b5bf6 : SUCCESS in 1h 57m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da2ce1b96b244e5b86454356fb7232f8 : SUCCESS in 1h 18m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87483c928b5946ab9f6b8ce3b2df03ef : SUCCESS in 2h 28m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/249866b876614674a4266f4f082bddad : TIMED_OUT in 3h 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3a1d8cea344dbd94ae066c0a2e6e68 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/754f7f8aec1f424aa17ad89cf4b537c3 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfc80124b0314b47a1c2641c9e371f06 : SUCCESS in 2h 38m 37s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/5e99077121904db293061f4091c7119e : SUCCESS in 1h 49m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ec62712baf2ba97a29bc16de86b599c1d904396c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 22:50:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8f3979de7c6c66bd476a83f8e25edbe16c4c93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 03:20:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffd83a98608848a79ffd377e3147bde9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24cc2fa3b656458ebd496ad3bcb08926 : SUCCESS in 8m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/22c04f8eb8fa475fb0c420c35261a578 : SUCCESS in 52m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2636c0984553444e8a68389f7ed29bd8 : SUCCESS in 1h 28m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0ad0d6e35c3e434891e4639a2c26f9f0 : SUCCESS in 52m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e652c5eaed74ec7a625f793a60f689f : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8f6387d2d0b84f6899aab9d0882535dd : SUCCESS in 53m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004bb10383cf4c59bab34df29eb5c9a6 : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9494300d1c3746c59d3ed86f821e7f99 : SUCCESS in 49m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0822006461f34036ba20f1ac419b56fc : SUCCESS in 1h 14m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/555b86a316ec4db29e551a93a8969e94 : TIMED_OUT in 3h 06m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b3e32a53d2746f3bed8bd5eb9fd15b5 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501a1893d5584de4bcdf83fd90a79dec : SUCCESS in 2h 18m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c0a45648c844cda175c3ae7d128e0d : SUCCESS in 1h 07m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76b1e05222d4c5e9af75e425f6f7839 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/396900ba74ec42209cd4ce0e683a4b03 : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4f07ebf3e4a4d84993712f392ed35b2 : SUCCESS in 2h 37m 10s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/5d149a70a1d04a299b2bbbf303c51809 : TIMED_OUT in 4h 25m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2ae55bd1bf19419eab6e535e501957f995673cb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-25 09:39:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dcae68e8ac779a5a781373302ec678664006b183","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-25 09:58:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1187bb24e6f6fcb628ab8af832314ea45139ade3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 12:05:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b87d74d9e1246198b585a7639493388\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9102a1fba1364cb1b4f6220f78dd5b34 : SUCCESS in 10m 33s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0811fbac126d491cbd0f30c99ef629c6 : SUCCESS in 52m 36s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3868558d915e450ab4531f23ff1bf2e7 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7f179fa30cfb445ba90dd523a26d3e93 : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b63b01f0e0284dd5ab73ed0e13e980a0 : SUCCESS in 55m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52c1b42adc774d8e8eaeca0f92b791bc : SUCCESS in 58m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a94597860a114b128893ae8cd086bda0 : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/363f4aa169ba4c6489da92b2ca5fe606 : SUCCESS in 1h 35m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a1f4fdbcee684a789c288ee361776754 : SUCCESS in 1h 10m 23s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/62c5ab5d88b44cc483010ffccd414526 : SUCCESS in 1h 52m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef2c7eaf94147b5b5d085c8b96ae06b : SUCCESS in 2h 14m 45s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e39c11c4ecd343689584ce5a1e5e5125 : SUCCESS in 2h 21m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2a16004cbd24643add9801db46aa418 : SUCCESS in 1h 17m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/459cc0a247c94f96989d5a8a58a4f847 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b3242feb3a54df79ed9ece3ad47d988 : SUCCESS in 3m 58s\n- openstack-ansible-deploy-aio_skyline_yarn-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11cf983249564ae68df7ebba35958265 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_skyline_yarn-centos-10-stream https://zuul.opendev.org/t/openstack/build/6b93def1e8694376b5d68c07cd35a200 : SUCCESS in 2h 04m 11s","accounts_in_message":[],"_revision_number":3},{"id":"f793e8c232d7dbf05a24668a0f9fe819b6f8344b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 12:08:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"67281401006ece4ade2ae851aba525ba15e573b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:55:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16cf8fbeebc45d2b9909f76d96744fd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1a2d1dd9b124bb78092210a7a8105e5 : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/051ccf8450aa41f7adab99c58d13f86d : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/25b6044a87244f938cca83c5ae5d9c02 : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9b6bcf03cf074c4c99141d288e4039dd : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/90166df4d764423a9802c89aae489ff4 : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d791f8ac359a4c698516f2331e1fe504 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1842be5dca8340bbaa3a2fb323218434 : SUCCESS in 58m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18fa30ad58fe4f37871d9419a9d6c0d6 : SUCCESS in 1h 39m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb34338c7604532b9e5653893b17f08 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a037a125f004e8184bf5a1fe97d10ba : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c6bf4331bc22a5074c13987689e6ef36cc80fe93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:55:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5201966cd35d11637bc059d297908eeafd42e573","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:56:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6164ccb5750b4e3780fe515cbd3b06ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a4bb1e74ba4f6f960a1b24f4c263df : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb7d68ddabb64d135177b358e009c3a50ce05857","revisions":{"88943c900b4fe189813c0663428560f194451e99":{"kind":"REWORK","_number":1,"created":"2026-04-20 23:16:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/985595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/985595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/1"}}},"commit":{"parents":[{"commit":"d4c27ce8186ec9e195458870bca809648da5bf41","subject":"Merge \"[doc] Add configuration guide for Skyline install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/d4c27ce8186ec9e195458870bca809648da5bf41"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:16:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:16:01.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: Ibf41cf2d902b239189f8647c9e67f654f6238d38\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/88943c900b4fe189813c0663428560f194451e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/88943c900b4fe189813c0663428560f194451e99"}]},"branch":"refs/heads/master"},"6a77ada761814541932e3c7f3531119f98e71a8e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-19 21:38:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/985595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/985595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/2"}}},"commit":{"parents":[{"commit":"f2c95a07e13c4f001edd6f3f5270df1f36885b94","subject":"Fix Object Storage section for RGW","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/f2c95a07e13c4f001edd6f3f5270df1f36885b94"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:16:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-19 21:38:28.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: Ibf41cf2d902b239189f8647c9e67f654f6238d38\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/6a77ada761814541932e3c7f3531119f98e71a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/6a77ada761814541932e3c7f3531119f98e71a8e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"88943c900b4fe189813c0663428560f194451e99","theirs":"f2c95a07e13c4f001edd6f3f5270df1f36885b94","contains_conflicts":false}},"fb7d68ddabb64d135177b358e009c3a50ce05857":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-22 10:02:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/985595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_skyline","ref":"refs/changes/95/985595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_skyline refs/changes/95/985595/3"}}},"commit":{"parents":[{"commit":"cf55e13675be92d6806988d9ec81dcedee2c292e","subject":"Change skyline_pki_san format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/cf55e13675be92d6806988d9ec81dcedee2c292e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:16:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-22 10:02:15.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: Ibf41cf2d902b239189f8647c9e67f654f6238d38\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/fb7d68ddabb64d135177b358e009c3a50ce05857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_skyline/commit/fb7d68ddabb64d135177b358e009c3a50ce05857"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6a77ada761814541932e3c7f3531119f98e71a8e","theirs":"cf55e13675be92d6806988d9ec81dcedee2c292e","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
