)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~654378","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I6a52f1eb46ad17c7a2a3cd5c3b1b4f19687aceae","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6a52f1eb46ad17c7a2a3cd5c3b1b4f19687aceae","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:16:14.000000000","updated":"2019-04-22 12:22:03.000000000","submitted":"2019-04-22 12:22:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654378-1555935723608-332a2ac9","meta_rev_id":"2883ebd7a1d29abf145c22a3d084c4b552836b85","_number":654378,"virtual_id_number":654378,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:52:31.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:52:31.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:52:31.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:22:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c249803e8d6c84894c3bdddb8d0ede30e6294fc8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:16:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f0fecedf2226bb25f04e8f9ef832a5dc0bc5ac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/654378/1/check/openstack-ansible-linters/660361d/ : SUCCESS in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/78/654378/1/check/openstack-tox-docs/f9ac71e/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/78/654378/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/dce889a/ : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/78/654378/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/9c4a66f/ : SUCCESS in 1h 03m 59s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/78/654378/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/264f5a7/ : SUCCESS in 56m 28s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/78/654378/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/2355d1c/ : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c4ee4b8999c11c4d0b26796c869ac5adf7383717","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:52:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c59cb0e21affa6962ef3b22aa7428932a58a2c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:52:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f335df472bbc17763b45234ebaaaf4fd4b6ab28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:22:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8500fcd37b5e31107b74a7fb2a0a731aa8bf448e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:22:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/654378/1/gate/openstack-ansible-linters/8e67fa0/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/78/654378/1/gate/openstack-tox-docs/980a971/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/78/654378/1/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/3aa4992/ : SUCCESS in 1h 21m 09s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/78/654378/1/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/dba0d3f/ : SUCCESS in 57m 30s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/78/654378/1/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/b9b6962/ : SUCCESS in 58m 31s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/78/654378/1/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/7093023/ : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bae76c0d14f39cecfcf85910731abddf9449f13c","revisions":{"bae76c0d14f39cecfcf85910731abddf9449f13c":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:16:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/654378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/78/654378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/78/654378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/78/654378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/78/654378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/78/654378/1"}}},"commit":{"parents":[{"commit":"6acf88aef7ddc01433e4947ed8ce18ecaac90f61","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6acf88aef7ddc01433e4947ed8ce18ecaac90f61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:12.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6a52f1eb46ad17c7a2a3cd5c3b1b4f19687aceae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bae76c0d14f39cecfcf85910731abddf9449f13c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bae76c0d14f39cecfcf85910731abddf9449f13c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
